)]}'
{"id":"openstack%2Frequirements~764637","triplet_id":"openstack%2Frequirements~stable%2Fussuri~Ie1569a24be5bd535a5868ef2f600d366207c0e95","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"Ie1569a24be5bd535a5868ef2f600d366207c0e95","subject":"update constraint for tooz to new release 2.3.1","status":"MERGED","created":"2020-11-30 11:18:05.000000000","updated":"2020-12-06 08:13:02.000000000","submitted":"2020-12-06 08:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764637-new-release","meta_rev_id":"f8b3dfee5889140b7c55d814500389143a8e95f6","_number":764637,"virtual_id_number":764637,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-06 08:13:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-06 03:39:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-06 03:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 15:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-30 16:43:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-06 03:39:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"39f9353ac179f339a84123a7104fe92e8a6909a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-30 11:18:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70fbc32fb09dfec58c01d699d460424347bca317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 15:40:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b401ca24b7c4f12993c9a44069b1f84 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bacf041c91ca4210923bdd8aa36d1b44 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c277565dd104a13841fcee634559b99 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f26e8d271c04e6d92d9be8933fa4c51 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21696f9ad5244153bb5d88d4eb6ffda9 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b401c3ebd2d74c9c98ab227dbb21c2d4 : SUCCESS in 9m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/590d5f2cb4d1434ea86db79f984bba43 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/940a494b4f934e819fe5d6e91589d1fa : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c92e941d7b2a432ba3d2ad630995d0e4 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2022df172984dc8a538ee9524a7ddb1 : SUCCESS in 6m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/551089650b054c7588fd325f9fed7ab1 : SUCCESS in 12m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/041f482b1aa641d2b3d24e77eb745a07 : SUCCESS in 18m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/138a6562d84b40bc9b8adf199c379a89 : TIMED_OUT in 41m 13s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/dd4e70fde2ab4707b5b86e9203732ec6 : SUCCESS in 13m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cec7864fb3164924849dd077b05cf0c4 : SUCCESS in 18m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87a49e704ce74182b86b596487fd6b26 : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/fdd2a6f92d03408aa16314f049fa38a3 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6583401908764ac7afc1441ea77bb5d3 : SUCCESS in 7m 55s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/60cf8890c2884820b30ec1be929ccfd9 : SUCCESS in 10m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8b4bc0ac834a499183e168b921b2c784 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6186afd8c5504790ae13c9af7d41bedf : POST_FAILURE in 17m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9e1a858e9994206aaf9b942c8434e67 : SUCCESS in 1h 25m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3075bf568b4e4e8baffa4e99485d9480 : FAILURE in 15m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/46f10735bf6043eb83bbb69ddbff546e : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"c7b802737689a1c0ca198de23dc0778df88fd00b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-11-30 16:43:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5584937885a656b6eedea0d24b6957949daf599f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 17:57:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ef00a240404579a67b78481a4df52b : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e570fcb97f8b41129938d19ea3363430 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5563a00487a44e5bbd27e370853262a4 : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4056de05ea3841b0ba20d2980b55c563 : SUCCESS in 8m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d0f55d480ce43a9a4ca8244aa5cbd78 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c3e2bce8b31d40479bcac440384a6d59 : SUCCESS in 6m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/41f204cbb82f49baba322c8b3ee46605 : SUCCESS in 8m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/07d23bcb6c00460988f5cb0c983f3b57 : SUCCESS in 4m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/14201cab2c4449dfb895909a45eedd09 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4afe4cd01de94fefaa3d7a6ac33689db : SUCCESS in 6m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7bf1b9ec69cc46759222647334bf53a6 : SUCCESS in 13m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3caf7992d321498c9e14fb5e92c69898 : SUCCESS in 25m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ec30fcae61a44841a26a07ed6e322ba3 : SUCCESS in 14m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5a27a424cdfb4177b3023e6ba5da94be : SUCCESS in 14m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f8c6d4ddb1af47f383c801f365beefc6 : SUCCESS in 15m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/98cdd1dbda6e4e9284adf3eef4286d04 : SUCCESS in 4m 39s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/514df12e2a78426cbce03e8ae8ca8eed : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86f490aaa0c54c6e9e19c0141572987c : SUCCESS in 5m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f0742f6b952042449af5db1b8f0366f2 : SUCCESS in 12m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1873803695c041d392e8851ed925f802 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cf720ac715944d3802f8be0f17ba031 : POST_FAILURE in 9m 54s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0abfa52ac62a4c7baaac012e519bcba5 : SUCCESS in 1h 11m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/063479598e5045c3b1816a3447fd19bb : FAILURE in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/2c13599e2e3e4e4d960343a41f14ca6c : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":1},{"id":"428d0231869e204c47179998b733b947c71d1046","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 05:54:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f60f118ea7d76607cbfbad5c6b1d301747e3c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:05:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57829449adb044c4ab914c49ffe436f2 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe77061c26b046758c93fa7da7fd63ca : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9584dc6be1454701b343d2acb3dc1d67 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/521dc58658e3492392aed0bbbc7df05f : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d18673875ccb439c8648efae526eac51 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/44131fe0e05e4e92be3a6592ddf714e4 : SUCCESS in 7m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4b6f41fe7cf24936a2dffad4218f1f7a : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59dc2af60ecb459d80fd3a0d3af7020e : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/eaf73e6d53eb45eba4a1189d84ee5ad4 : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8506beaf977c496bb7ccd858a7c7dc72 : SUCCESS in 7m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c63ec495fc5d4358a65be4d21ec9c983 : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0d2d757728534e438961bf9aff38b1db : SUCCESS in 31m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/533d036b41634e649a886c82eb144cc0 : SUCCESS in 17m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fa39a6f7cd60429fa8418d3a6d02f7ee : SUCCESS in 20m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d7f34f07a33b430796599608fd5c2e13 : SUCCESS in 19m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6dfaf7f679a64c94a56141cfa398a075 : SUCCESS in 5m 19s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/a625afd1240c4dc3b6eb8ef3696f3250 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ec3503e4f88462791a96f531a547fcb : SUCCESS in 6m 39s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/734e004e40a24267965e16df1d325896 : SUCCESS in 13m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a5ff9de87eb4c2c953599595746554d : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/732f2bc8341744bc9bfb24c7b71c06b6 : SUCCESS in 1h 10m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c40a2c64a68d47009ab777076d99dcd7 : SUCCESS in 1h 41m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f0e512027fd7476cb933078d0cdbd6f1 : FAILURE in 53m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/b063c65d07a04a03857a0d30055fc32c : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2877bd75377d11ec9ef9fe7a56797a39d4b993f1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-06 00:53:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8d5e843805d74964787e5fb9b6cc4f53a2286243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 02:30:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b601aff2ca411d966af04c73b8f49d : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7d8e6fe785844b3a57460997e76b1a7 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/80dc7ec0c8144201bcc69ede4bfad2af : SUCCESS in 3m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/751648a932cf4251bb1b4d5d386319a1 : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62d5ad26f8c540f4aa6eec1c6544886a : SUCCESS in 2m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/662981fa87bf4268bfcbbde5f8638b3d : SUCCESS in 6m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a0fb957458a485a9e12ee0a92d4f25b : SUCCESS in 8m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7732a91d7dc74ef19294a13ed2cb7fcc : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b2e4eebcd6f455080ca58067622d84c : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/00a226449f2744658aabebbf39de0adc : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/503fe6b9f3264d3e9c292ef6c6713788 : SUCCESS in 34m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a97d183519f94d4286a3b5d0f93fcd40 : SUCCESS in 28m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/53d531685bb548c09cf50a9db0558b18 : FAILURE in 22m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2963d69b8404b4aa2909e6c0c734b55 : SUCCESS in 15m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce052f7e99914c649d94cea2d1315f09 : SUCCESS in 15m 07s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ed39e127fa664d759dd92cadf6eb9d03 : SUCCESS in 4m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/119f8ff3faf6410cb7e9524c989031a5 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eff681a3188747dd93bda8085516040f : SUCCESS in 6m 49s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/82a9ad549dbc47a68860bdc9d62d7fb0 : SUCCESS in 9m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f165a465414d43dfbb8f82fdbb450d00 : SUCCESS in 25m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/158ce39413354f228586355c03f866d1 : SUCCESS in 57m 24s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a72570972de34279a3eb74f05ba34ade : SUCCESS in 1h 27m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54c2f3d7a9e54effabdb07524bbf0914 : SUCCESS in 1h 31m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/5ff9e2bf3db04d379a12617af4c9e667 : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a79cb65cd2622f2886868aad678e946fdf5615c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-06 03:39:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"46d374ae113a23bbe99186bb472c4176d3e5f0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 05:08:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e6ce987d9146ddad45a4cdc049daa2 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a48f6116fa1741a18b3df25bc86dbe8f : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34e28e3e97c94ae4a5ae93ac2be74fc0 : SUCCESS in 6m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b0d8dfbd7b546ec809fff51167328b3 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f6117001e314a5fbd7a4043e58fee7b : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e148a1356ce4accaa86843622495f2f : SUCCESS in 7m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff3dce815e01400b8ed34a4bcace9ffa : SUCCESS in 8m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b199f040487462a99eb572b9a45c44f : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9b6aa32ef0e6447a8a7f1a05ac8c74e2 : SUCCESS in 8m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3a9a7c4842c4bc68bd702be73533acd : SUCCESS in 10m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/864d0920a13d4a65a3484b78931ce904 : SUCCESS in 13m 39s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2df95e0b31734fb39e55fe8e9f49cb31 : SUCCESS in 19m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5519f7c8f76e4299a974a14c719de873 : SUCCESS in 11m 54s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6fe3b94f6e6b4235a94b5f290ca6790a : SUCCESS in 13m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5778c3a6be5b4c1a8d8646976faa315f : SUCCESS in 20m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c41dc059614480d9b25a85eeafce3a0 : SUCCESS in 4m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/8c92b366eb0f4b81ad1ceb36dda8a3a2 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5bdbd363b51c40c3ae96cb60ead9b004 : SUCCESS in 6m 13s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b185b6a818924b1188b5ab203a8a8d72 : FAILURE in 13m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c3a7f71b5ad14863a7ca2f05974287c3 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ec62345bc8946bc9b9969f9f73aac54 : SUCCESS in 50m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/395e8a51923448d9906a9b45c96c24e4 : SUCCESS in 1h 22m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac5e6c0f16bd45a6951e36e51fabf224 : SUCCESS in 1h 14m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/237dea46e17d4ec1baeede0bbce810c3 : SUCCESS in 1h 21m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e6e2859acb70cbf14b43316c7dcbf148810a03bd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-06 05:11:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4575cd211a1b15ef4b455b6fe4ec1b728ae8caf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 06:28:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78f970f3db94a299e502be0e8ba26f1 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b854dc2fa7c1433ca395dea96a7240bf : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17f659604ae949ea9abb3a4c151a460b : SUCCESS in 5m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3fcb4774ec5a44948e8edcb0af3eeff2 : SUCCESS in 5m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/620afbf0dbb04bf1b5b73791fcfb5f6a : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9d31a21ad5ca4978a35ad41d80f2bc34 : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/68d800144e284cdfb0db11ed5e277fdd : SUCCESS in 7m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/12923a82f1ef49e488cee301a819881f : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d4defd94b7d74674ae1fa5f195f7b74e : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97f96e65e42c498193dc72e846449a07 : SUCCESS in 8m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1defe4a3a2f4b79a98a0b12305a43bd : SUCCESS in 20m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a33c248c106c4d1e82e5a95d895b4dad : SUCCESS in 18m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ce8ba9b1d3f147dab0b3f8f75a8106d3 : SUCCESS in 11m 30s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fa411e83af764c1a9f7f8e8cad22d70c : SUCCESS in 14m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2cbadb679fb34637a03256b9f8cd1af1 : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1d9e413bc057434788a1a1237c4511c5 : SUCCESS in 4m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/9fe1031f455349f58aa69fe385cb5707 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4d824ed404d446d8b850f31a2d1ad2a : SUCCESS in 7m 31s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/684ad1ffed014b15b2d98c52f6fb9ae5 : SUCCESS in 10m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/296587bba0184623a5c0ac38ba75f390 : SUCCESS in 15m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64c3012601094eb8923524c59d1390fd : SUCCESS in 45m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d3c7a8233e2457987e1cfd76ab23c7e : SUCCESS in 1h 00m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f1e605c756b40da849c7c35c2d1e6af : SUCCESS in 1h 15m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8230433c126d4d2fbce4030bf0bbead3 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":1},{"id":"20008fd707cbd4356824b5809e860c49eb042d83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 06:28:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf65b42fb2493bc4bee913b93c249c3cbb151dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 08:13:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bb55deaf09a4bae99419a42dfe02251 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/027e12a39379439383070daa2a1d24fa : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57ba769244f449c1900534378078b600 : SUCCESS in 11m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/64bae236097d473697ae48e0d8241858 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c126b144335430fb10a9db0b67ead14 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fd7e75ca2fab4a1d9ecb079236ed8505 : SUCCESS in 7m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2ed87df2a9541a68758f8e4f4e90c38 : SUCCESS in 10m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/91c1389218374c5da0e0690a9537513d : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4e07bb4c8842434ca96f28e1e00d139d : SUCCESS in 10m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4bcc17620a204fb5bf34f146f5a655cf : SUCCESS in 6m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/770f2415410949eba02d4cd5d9192e81 : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fcfabca625b84a9c8a5ca65f9e55f39f : SUCCESS in 28m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a296b4d7127149f7be3f8e18726769d0 : SUCCESS in 19m 50s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/4270d93046a34fdba07fa44a49c432a7 : SUCCESS in 12m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5830b5cbcdbf4ab7a1f4faa36cb820d4 : SUCCESS in 17m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fd99872fd752456e8c1e6262acb7dbc2 : SUCCESS in 6m 18s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/1de590926a0442c6bbf9a49221301b2c : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/21739a4ce792415cba523838104dc7ad : SUCCESS in 10m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b10b01c40b0449509a7126d7b7047286 : SUCCESS in 8m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c39cc2375eec4a69a3b341bf1370cfbc : SUCCESS in 9m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18cc38a1ec19466f86da550e9136919c : SUCCESS in 46m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b827bae71777431e8be2f1994770af0a : SUCCESS in 1h 06m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/023f8810aaf74c62b7db6579dc086f69 : SUCCESS in 1h 35m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/4d2af7d4bc9648b1b42eb83bf5d35000 : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f8b3dfee5889140b7c55d814500389143a8e95f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 08:13:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f734164fab12efb1e6748d2bf6964cfbd0cee3ac","revisions":{"f734164fab12efb1e6748d2bf6964cfbd0cee3ac":{"kind":"REWORK","_number":1,"created":"2020-11-30 11:18:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/37/764637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/37/764637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/764637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/764637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/37/764637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/37/764637/1"}}},"commit":{"parents":[{"commit":"bccd484efd26f493c00e5a41bcca29ea5a4865bb","subject":"update constraint for oslo.concurrency to new release 4.0.3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bccd484efd26f493c00e5a41bcca29ea5a4865bb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:18:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 11:18:00.000000000","tz":0},"subject":"update constraint for tooz to new release 2.3.1","message":"update constraint for tooz to new release 2.3.1\n\nmeta: version: 2.3.1\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: Idc2200d09bdfd473153558a02bb44f17422d6f4b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie1569a24be5bd535a5868ef2f600d366207c0e95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f734164fab12efb1e6748d2bf6964cfbd0cee3ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f734164fab12efb1e6748d2bf6964cfbd0cee3ac"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
