)]}'
{"id":"openstack%2Frequirements~764809","triplet_id":"openstack%2Frequirements~master~I45806224b4397568b054932dcad25c9d5e9020c3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I45806224b4397568b054932dcad25c9d5e9020c3","subject":"update constraint for stevedore to new release 3.3.0","status":"MERGED","created":"2020-11-30 18:03:20.000000000","updated":"2020-12-01 06:42:43.000000000","submitted":"2020-12-01 06:41:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764809-new-release","meta_rev_id":"5cb27bc7ff4d1d8254d6764af19c5aebbed6d9b3","_number":764809,"virtual_id_number":764809,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-01 06:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-01 05:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-01 05:03:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-30 18:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-11-30 19:29:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 05:03:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ee6c85b0331f8cb8b87bb4406629f7bc75e67dd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-11-30 18:03:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1459f51895b5e954bb9ae80379cb0435c5743fe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 18:19:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c7def248ddcd4360bff577fd691297cc : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bc4b26a0a389bf89fde5bce545c5a4576a0b5d1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-30 19:29:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dec70ca4b8f478f95219384c2fe5283 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b788b40a1c0b4fb79169bb4803ecc61d : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d767922e611546148cfb574cec1c09f2 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/112ccfb8ff134fc7b35a216da591bb08 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/753d0e51b35b412085ea9f27c10f0e8d : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/636b36ce51a1489aa2bb5d17e298528e : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b95d3d81b6ac48adbb8c5aff320d431f : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/faca553c4870434abeabe270223383d3 : SUCCESS in 7m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b09eaa0a09214f11bf9760b53063e32f : SUCCESS in 7m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10f85875b52c4b88b62a42da84eaf788 : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ff0f1a75e3114f73aad4e17c2d612eda : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f48d59fca0fd4d4996cca7be15909aea : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f98aa16ebc348e18a5a6c35d2543d40 : SUCCESS in 6m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64fb6216bb8b470f808c4b9a06e7f19c : SUCCESS in 6m 50s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ea379f1f945f402f98ede531c1c45530 : SUCCESS in 13m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/13aa7c0bb176409282ef36a59ace5cda : SUCCESS in 24m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2e51a08609b94f00a140b6eedc67954e : SUCCESS in 11m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1adb9047887e443d905e7f9ba7f4d1da : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b48040bcb26c4ec78ad1069ec1b8d1bd : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/378201ab5108485789db0155ae558cb1 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e445959231374d05ae5d9dd96e585ef2 : SUCCESS in 5m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/625d9a12c29e43fe9f34cce7a8f56bcd : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7167161cb52640698ff995fcb0c1241f : SUCCESS in 58m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0ddf65f5e9bb4e54812e3546c2a7c44a : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7431b461b92bb45039e53619adf687458edfe611","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-01 05:03:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5102b66f0794754386a0282d907df01279d692fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 05:03:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5794e89d04090e151cd25528addb798a08ff99c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 06:41:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663fe1c3d50d4dab95826bd51ad82687 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082621c8e63f43f6a218486ba9fc09b4 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/013a697f2f0d4b73b11fc5bf6100f0c6 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fc960b665514881afeddc815b82d9ab : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c14679357be2406eaf0123a1af90551c : SUCCESS in 3m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c436cf6f21d4f06ad2594e1242d1dec : SUCCESS in 3m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a53b57650441473fbb0e4ca8c755e38d : SUCCESS in 6m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/013b45ba9bdd45668338c808483a1a84 : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ef620f9658744b8eacf2208c2f025e59 : SUCCESS in 8m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/596efeecd1c44d5bbe6146812118990c : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4ae96fe271c42d2922838c61ec79cfb : SUCCESS in 9m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28dcf82118194b8d8329f398d00035ea : SUCCESS in 6m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e3448b5884844e7686696a6ebed03de9 : SUCCESS in 6m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/77c4db0efe4344cab160f8d59dbf2d37 : SUCCESS in 14m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3efbf9596c4543f7872fbf18b705c4ae : SUCCESS in 25m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ccab1fb1f7841e1ba545434aeea290e : SUCCESS in 13m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc335330ae6d4b99a24b504a7f790459 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f069af0483c4944b9126d349092d9b5 : SUCCESS in 15m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/edce265eb3d94da8be49c690ba1f52f1 : SUCCESS in 4m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d10c19df823241078c2bbebf16cb4481 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d1356894cb74305b5b8ae41d3bd1612 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31c553cc316c481abd38a86fb9b667aa : SUCCESS in 50m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/288e536dd8124bafbf0c72b1f9493ef8 : SUCCESS in 1h 31m 09s","accounts_in_message":[],"_revision_number":1},{"id":"dc23ad50cc5844d440a8992b869c272cd829455f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 06:41:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5cb27bc7ff4d1d8254d6764af19c5aebbed6d9b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 06:42:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae4ee9e3adc4d87b17497d5df09a7ac : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7c767c079eef6fbb7ff338176e4f4d575cd39ad","revisions":{"e7c767c079eef6fbb7ff338176e4f4d575cd39ad":{"kind":"REWORK","_number":1,"created":"2020-11-30 18:03:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/764809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/764809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/764809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/764809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/764809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/764809/1"}}},"commit":{"parents":[{"commit":"f74b67ac0a5a42e7f933c5cc75201e835d3413a7","subject":"manual update because bot is not running","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f74b67ac0a5a42e7f933c5cc75201e835d3413a7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 18:03:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-11-30 18:03:17.000000000","tz":0},"subject":"update constraint for stevedore to new release 3.3.0","message":"update constraint for stevedore to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I41c36467f5f0e130ab6c40299f536a66d6991bc8\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I45806224b4397568b054932dcad25c9d5e9020c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7c767c079eef6fbb7ff338176e4f4d575cd39ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e7c767c079eef6fbb7ff338176e4f4d575cd39ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
