)]}'
{"id":"openstack%2Frequirements~764990","triplet_id":"openstack%2Frequirements~master~I165058a2ffe2e5236a7ed9af71ab879042ea7aff","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I165058a2ffe2e5236a7ed9af71ab879042ea7aff","subject":"update constraint for python-magnumclient to new release 3.3.0","status":"MERGED","created":"2020-12-01 17:56:24.000000000","updated":"2020-12-02 05:00:42.000000000","submitted":"2020-12-02 04:59:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764990-new-release","meta_rev_id":"8c58b39ab24e7d965bc13ebc8a52cae77e0a3e42","_number":764990,"virtual_id_number":764990,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 04:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-12-01 17:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-12-01 19:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-01 19:28:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 17:57:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-01 18:29:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-01 19:28:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-02 03:18:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f01ed2193ef0aa70954fe1c14049683ceeb14bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-01 17:56:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1f2d5e1ffc6dc07cc5affc3024d697c3910b189","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-01 17:57:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"82c11259abef98539712197a58ffdcb33efc25e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 18:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dffb1c660ad641afad6a018317b32fe3 : SUCCESS in 18m 18s","accounts_in_message":[],"_revision_number":1},{"id":"328d11aa1f2b3d11d011053ae44646dc978f6be8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-01 19:28:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"812041098da8b140a44668b095ba3d8535025048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 03:18:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e4784b7be944469f5797347dc003cd : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e16fd055f861443eb2cd23d430f1635a : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f42b2b85f500479c8f85374f16fe0eed : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4684d932be24ea0bb62952cde3f1f91 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ff4ae2580f74f199fe0f9162c4f677f : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/009616ae89204613a2a346522ab4f029 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2a78bc48a48492e949276925a8f6c09 : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3037b3f03e704a7f86359360852bda05 : SUCCESS in 8m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11f34d25f7c74d0899115875a21c3d8e : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/55365b774f3e4ff2a9eaa8dbb4f8c2bb : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3804c57d5a35438da46d354e9c02ba94 : SUCCESS in 5m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/83ef770fdf984767ac627738771221fd : SUCCESS in 10m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/24375bb543294c5e8322673d519ca3d3 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc2176d32eac465ba6d7b356fa7e33c6 : SUCCESS in 13m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/37d3dc14d2274a5ca386eac7241a3331 : SUCCESS in 18m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ecc837e82d64394bb8795a35178c368 : SUCCESS in 34m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be3c39b253a14c008197517799fce0a2 : SUCCESS in 19m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6183e46bb8e74c5bb46a9bdb4697b340 : SUCCESS in 5m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b66114318aa54ca280e4c73aafea4aea : SUCCESS in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ec57a6c378bf4094bd78af81a6c2eec5 : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ee2e5bbb4f34306bbbb827de8d5d52c : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b77aae4b052144e5b496ff308129df48 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f7c6e4fdcad4e1d988a99fad75abf8f : SUCCESS in 1h 04m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4405ff4a18d64fad8fcdb13ba562b04b : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0ff96c9c813c8099f07d08eb5734e8ae4773497b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 03:19:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1840835c051a3fd9c3db9b3ccbced366c7f96f43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:59:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70678a1f745045e784b1c7cadb4c90b9 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/607ce85e344e486ebf7b7a42e625f818 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d03afaeb65141b8ada0b723d465e8f4 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b0668b33eb4135b81490d632b35fcb : SUCCESS in 12m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a082f0bb15784dadb5966b64917390d0 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbac452638a2494ab35a56341f711b4e : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b4055dafd8248d1b68a3751d562692a : SUCCESS in 8m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8cea0463712b47008ae980956f8f5e30 : SUCCESS in 7m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/572ae11874f0403280712e72a405c8b0 : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b394333632e47d188e1ec3068925ee8 : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/42e1e2a010c844d2bba7ed1879da0348 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6b54d8277a434f5b835086e3e8024aec : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7af8f0832dc40e29d6e312f964bb9a9 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c71fe50ee83418688e163ff7492bc51 : SUCCESS in 11m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/53a18ec87c78492da2bca001add84020 : SUCCESS in 26m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/09ff56d8771f46849a7fc0d1d3f112a5 : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fdaa316bb61a48979727f877f2e4798c : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83e05f77911149b58927293dcd0cb135 : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/548d241504f84f62a3ca4bdb20522db9 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86a14a93f21f48b58213fbae4b2a0783 : SUCCESS in 6m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/401cd98085bb4856a414ebbc23503573 : SUCCESS in 11m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c69dea1dbac4117aeaf4c0f71eba70e : SUCCESS in 49m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/931e8d6db58343d585d7cb89d55575b9 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9772b9f2430559465680ea30aa181d22cc887f92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:59:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8c58b39ab24e7d965bc13ebc8a52cae77e0a3e42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 05:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96219c83335b40ceb38f5fac5cce4eda : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6aff26adba9eb15322aefaffed9d5c63212c1c97","revisions":{"6aff26adba9eb15322aefaffed9d5c63212c1c97":{"kind":"REWORK","_number":1,"created":"2020-12-01 17:56:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/764990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/764990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/764990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/764990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/764990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/764990/1"}}},"commit":{"parents":[{"commit":"d7f43d40ada612e606443b1d819819dc19f48a60","subject":"update constraint for python-tackerclient to new release 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7f43d40ada612e606443b1d819819dc19f48a60"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-01 17:56:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-01 17:56:20.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 3.3.0","message":"update constraint for python-magnumclient to new release 3.3.0\n\nmeta: version: 3.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I03eaef53b8ee0c3178577f2c4447b946ed270715\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I165058a2ffe2e5236a7ed9af71ab879042ea7aff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aff26adba9eb15322aefaffed9d5c63212c1c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aff26adba9eb15322aefaffed9d5c63212c1c97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
