)]}'
{"id":"openstack%2Frequirements~765098","triplet_id":"openstack%2Frequirements~master~I97eb8c526d371f011ac9ef7d548a3b1232cce27e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I97eb8c526d371f011ac9ef7d548a3b1232cce27e","subject":"update constraint for metalsmith to new release 1.3.0","status":"MERGED","created":"2020-12-02 10:32:48.000000000","updated":"2020-12-02 20:22:21.000000000","submitted":"2020-12-02 20:20:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765098-new-release","meta_rev_id":"415fe74e0f67ca2130f6ea0a7e44a734f36bde6e","_number":765098,"virtual_id_number":765098,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 20:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-02 18:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-02 10:37:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-02 18:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 10:37:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-02 10:52:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 17:27:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 18:42:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9216dfd437c89b7c7b1a0642b236ede4886124ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 10:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9468c1b0ea324ac1f05eaef97ff69f045ab9c23","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-02 10:37:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ce7693ae1a113975f1ba9c850260db5a7cff5bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 10:52:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f8c15b2ef2d445f48124b3e0adcf6ccd : SUCCESS in 15m 39s","accounts_in_message":[],"_revision_number":1},{"id":"095a78fe705a7b23151715e239736f29c016f02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 17:27:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb0e087a291948bfa5a68c0bb1070536 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca848325a40403d97b14e2c9cc0e10c : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f04016ba0c744b7ba46998e1d21850c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220cb3d3472a45788193079b9e7b33ce : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61716980b7964231b95f2c0a1e485935 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/888412254750456987c0c32f57586349 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b1933fbd3c3448f99c45790bcc27844 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d3d3e894790485791de212f415c532b : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2850341dda3d48009d78d3ce8e5b5fd3 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/22f9fb33eceb436a8c5f7d26aaf005e7 : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d93a698d27a546649c907efca9690427 : SUCCESS in 11m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30f3fdcb5b65476d95fdca50de65a70b : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3d1c8076cbe48d09c5640a3eefc5305 : SUCCESS in 6m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f6289ab9d1b54a75998f00d936d19f5b : SUCCESS in 6m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1197d0ebf48b4f4d88758dff5e6919ba : SUCCESS in 17m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/049765ee68934775801e363bc1fd70fc : SUCCESS in 31m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3141e7ff5da746c893d807908b411188 : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0ca25c07cb0b4f9c8ce3ecfad884dd2c : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/156617e38792443a93f274425721ba63 : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/86da4f15e4f5416ba5f51564cbadfa1e : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a3ae8c3cb60434b99bf93e27a7bf01a : SUCCESS in 7m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/24148880403e46b5a04bca0e26ddf9d6 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d20eb747ed44fde8c951b7398c53c27 : SUCCESS in 1h 04m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34edc1c961b243cc94d862aa1a75d3cf : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"85a8ae32dcf4c5d2c19d7d505a648aa80c9316c8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 18:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb500147e70f5738fa5fcb59767844d30a3bb916","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:43:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"244a86dcd5a3c8f845418cbdf042e374a6c54690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:20:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951e37e804054b2398330b507a769a84 : SUCCESS in 9m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6353574b630444396c6a8c6103e0ffa : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b7c547314b040cabd7cac911467044e : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d04d0b91b964072a21730c627e78940 : SUCCESS in 3m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f8d5f57a4314f11aaaa74139dc3ce7e : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/91ef386174074707b2140e94857a70d7 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/73bbe2433cd942888278facc8cefe512 : SUCCESS in 7m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc13889794914d02b81e1609720fa719 : SUCCESS in 9m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b5d03004db484fc5a3c6048381f2755c : SUCCESS in 8m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/098bf19fcc3e4ccd8f1f1dcd9805d2e2 : SUCCESS in 5m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b473061b57da4c91b0ef69b80996dcb1 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72bc4f84de344e2ea7e2d2f96e02fd9d : SUCCESS in 15m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70c9feca4bc64876b41a281acee61142 : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6170ef000aa46948ce676f3995d0915 : SUCCESS in 12m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b155ed83b2014334aeefa57dd09e80d9 : SUCCESS in 25m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35164e6810514c46b045c9180c7987b6 : SUCCESS in 14m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/781384dab5ea400f87f3e6ff2b2b2748 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85d0fee513d142cd87bdc9f76c5a789b : SUCCESS in 18m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/de276a4b9ec3400899eecab38efb0669 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/581cd040daee4efa8126fa5e0beec83d : SUCCESS in 5m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3bf8f5f2bc3543d78c1c0c7b781f4759 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a23c0c5009ee4cfd95daefa7be39f12b : SUCCESS in 59m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f8861fda7d546e8abd936420aa4f3f5 : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":1},{"id":"146e370a782212dac781eb420a145ce6ae0d31ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:20:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"415fe74e0f67ca2130f6ea0a7e44a734f36bde6e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbb19d72e494c88a6f8594ee8f47356 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cb9b0c14e4f9d36fd0eb2f535c5d036690a1ab9","revisions":{"8cb9b0c14e4f9d36fd0eb2f535c5d036690a1ab9":{"kind":"REWORK","_number":1,"created":"2020-12-02 10:32:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/765098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/765098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/765098/1"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:32:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:32:45.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.3.0","message":"update constraint for metalsmith to new release 1.3.0\n\nmeta: version: 1.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I1ed118e982d7f4075b83377b2fd3d4945e013568\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I97eb8c526d371f011ac9ef7d548a3b1232cce27e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cb9b0c14e4f9d36fd0eb2f535c5d036690a1ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cb9b0c14e4f9d36fd0eb2f535c5d036690a1ab9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
