)]}'
{"id":"openstack%2Frequirements~765099","triplet_id":"openstack%2Frequirements~master~Ib6dff04e0743bb72b7296121c3065126a3cdca89","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib6dff04e0743bb72b7296121c3065126a3cdca89","subject":"update constraint for keystonemiddleware to new release 9.2.0","status":"MERGED","created":"2020-12-02 10:34:54.000000000","updated":"2020-12-02 20:23:35.000000000","submitted":"2020-12-02 20:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765099-new-release","meta_rev_id":"2416e8a7865bb3d4d874db03eb626e5b2e72ebb6","_number":765099,"virtual_id_number":765099,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-02 20:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-02 18:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-02 10:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-02 18:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 10:38:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-02 10:55:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 17:26:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 18:42:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8418d03dcab86496580d067f2c0c2c48e92d3829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 10:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cadc2cf9213755460b0c3479c9f990de371201b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-02 10:38:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fdec158bc0cda33a7102a338dc3bcb77f3ffa2be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 10:55:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2e23898b65c5483ab58eeefdc59e27d7 : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"444059b96f4cd237e01bfbb7c8a52f53c83318fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 17:26:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0eda556e4f34d3a857a742735c66450 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1462526c2349f9a3189257adb75b9c : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7123492ee679401fa4bf85756e8689cf : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/483778538562494bbc54cea0a6b04a53 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16ac7295127c4454bedf9898916aa50b : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd5667ab92ad49db9a766f227016a3da : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0239870a51e6400aa166154a14ecf1cf : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8ddb6e5f93694f14a24d80b8f5ed4987 : SUCCESS in 8m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ec02c31f82e742628b9a597244d271df : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6ad7ee2f6a884da9b9af434889d4d9ee : SUCCESS in 9m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/620b064e00784d0c8966f11481bb8603 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/262d2be0477845f59c29f156f2d08e12 : SUCCESS in 11m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af2fc14996b0445aa6c09f8129e6df5f : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16734cf65afb424e9c9caa582dbf4c20 : SUCCESS in 12m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a2bb2328daf48c096270ba9b0f4f127 : SUCCESS in 14m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/547b39d8669540459c0ce9ca2fb7bf8e : SUCCESS in 32m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/522c5902cbe940169b145a6b047315f9 : SUCCESS in 12m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/53b0536365074567ae45200a37acba87 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1826b19da14b43d6af4d43fc0355ec47 : SUCCESS in 19m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/12a18dff73004dd290ff4acbd0b7ba0c : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53ed11fe08d74242ad33d8ab4c71a3b2 : SUCCESS in 5m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4f1c88d53620455bb1289771b440b025 : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c802573dcbd64664927363e83179eba2 : SUCCESS in 1h 00m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e5d68e1f78f4ed0a5fcae5e8029ec8d : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8f5bef5af95f229522a89dcc0fd2b6c8f2990b0f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 18:42:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e1084b700b56a8c57217ef883755c9ac21bc654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:43:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0539f09d1e44aa21409bc5ba31c98dd7bbe6dfa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:20:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9639d2531a894e9185623e7072426c04 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277d537691924264ad607d89590cebed : SUCCESS in 12m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f20969993d57422abeca12ae6bdd2f44 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ffb28a5355347fd8f0e157d56dd08d1 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dce2940b9ba4ed59c95a409ea3d4790 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71c7024b5d7c4021a896665f681df43d : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15901e75ee0c419ea74fbb01ccf30e94 : SUCCESS in 7m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c31efe5a2c2d433fb5f6a1cbfb86ba72 : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/37d0af11ac0d4a0b9d2e8d7707ab3ea8 : SUCCESS in 17m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/245ba91dfbbe4647b965b3252c281398 : SUCCESS in 4m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8ae1e787af26403182fc1e89b7e791c8 : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/41a460b4e7e649888275f0c1f4991e0d : SUCCESS in 8m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3f652d1dfd14668abaa89eb5c3475df : SUCCESS in 7m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b8b7aa2b51fd43a5ab3c2c1a5ab5f3cb : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b4250abf12a74e9496d36072804b7992 : SUCCESS in 28m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/18bbc350a4d147dab5cf68d76504531c : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1aff6377b64c4fc19fb363a2873fd164 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8070b99aa004e0c85abf34f3ddc508c : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1baf892e646f43cf8ca982f7a2eed9ee : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57c01b9e2e354fd7b3ec3ff51235c3c8 : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/42a91c345f7443de9522248c86ca2ed7 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5f97616e065445081a5db1c8dc04c17 : SUCCESS in 56m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/586e32a14759444a8138b158e5f4944e : SUCCESS in 1h 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"03c177ee011c4a413490e0e72ecaf183a34cbc9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:20:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2416e8a7865bb3d4d874db03eb626e5b2e72ebb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 20:23:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66c9701710b43c2a69f21f78fa16782 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"184a2817458fe12f48315210722f082e7d999200","revisions":{"184a2817458fe12f48315210722f082e7d999200":{"kind":"REWORK","_number":1,"created":"2020-12-02 10:34:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/765099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/765099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/765099/1"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:34:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:34:51.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 9.2.0","message":"update constraint for keystonemiddleware to new release 9.2.0\n\nmeta: version: 9.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I10e2fac4fdef97e95d4d10a2da9bb0dd0ad483af\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ib6dff04e0743bb72b7296121c3065126a3cdca89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/184a2817458fe12f48315210722f082e7d999200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/184a2817458fe12f48315210722f082e7d999200"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
