)]}'
{"id":"openstack%2Frequirements~765102","triplet_id":"openstack%2Frequirements~master~I8fb6d76d327fe39101f38833f8653c0280af2a4b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8fb6d76d327fe39101f38833f8653c0280af2a4b","subject":"update constraint for python-masakariclient to new release 6.2.0","status":"MERGED","created":"2020-12-02 10:43:08.000000000","updated":"2020-12-03 10:12:52.000000000","submitted":"2020-12-03 10:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765102-new-release","meta_rev_id":"9666b174f13ae84e9916e2a0f8868974c2d1a083","_number":765102,"virtual_id_number":765102,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 10:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-03 05:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-02 11:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-12-02 11:39:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-03 05:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 11:03:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 11:06:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-02 11:39:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-12-02 18:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-02 18:43:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-03 05:53:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"141550d95f32ac05b5e1a8e9a0138ddda9babac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 10:43:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6594674550fd9e6dd7f9a9e930ece32f8913ad86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 11:03:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d292dd3d4c34f699a586b9697676559 : SUCCESS in 15m 32s","accounts_in_message":[],"_revision_number":1},{"id":"762e45010d67d32bac9bf7b4ebe1640e9e306a72","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-02 11:06:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"542e6b0a91616412e2c76a4476dd9e62aff31fde","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-12-02 11:39:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9b884f7d92989dbe18d8a1d1bf65c2de5e67dcf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:09:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dacf58d12ee4910acf7a41b0dc87444 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cad77141dcb24228b12aa70fdf10b998 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cc72c277d7845c1ad5c292a8ecc3107 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9708f52c6ce343c68608e727a60027ed : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b47e7604be2b4e5a86e0c09e83a106aa : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c78e137e99a4abb9f908a97a7c4281f : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1246d90228934585be01f392643b8c00 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/468abbbf5e0d4f1c855e9d2c870dcc99 : SUCCESS in 11m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9080ef1086d644cdae52ad5bb5bf1b40 : SUCCESS in 7m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8860088471ff4e67b5586ebea2460025 : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/50e4108802ce449a89739545a0b5368e : SUCCESS in 6m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/63136ff1f3cd4883a4a5387778003090 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f3f05f136df48c8b7def4b0f78466c3 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86d827a3a026403cbf58da258f17c3c9 : SUCCESS in 13m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7f68ead56f214e918f43b68033b76dd6 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b2e5de402dbb4fdaa54cead22cada418 : SUCCESS in 24m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe2ec2442f2045dcb7aa1abcdc78953f : SUCCESS in 24m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c49aea49038418eba17efb7962839fb : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a614af8681e4d539940a480983b044c : SUCCESS in 25m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b2b03594a75e431d81d5e4ec37d14b6f : SUCCESS in 9m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/503085ab4c154686b3d861e635160ebf : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c764595035c346f99479f363eb731bff : TIMED_OUT in 41m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/815af9e6bd384b1db2a37d6f1c095dd9 : SUCCESS in 1h 13m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d4b247e161734c36a84a5538f8f7125c : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4323c0cd7ed7715a21d43e8aeef7d65943659c65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-02 18:43:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47cab763fc5154207ba0773fad2abe91b2acff52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 19:12:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/56331ab349c64aa58adae0dab6639132 : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":1},{"id":"62417b9f6e64a9bd7225fb3e1b1cfeaacd771c81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 02:01:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5b6f0d4c1a4dc2ac9827181df5dfc8 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c66304f698b040edaaa7e774dbd20830 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a80132cb139a4eac86220e24cc96f359 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e8e9158ed604b0d9d75f99e360772a9 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1dd7a602ea394ac98a79819406b670b1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6102e3ad56d45adaa6eca3743bceb09 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f88430ee6bad49bea640f8d27b4056d6 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0987df5323b7448ea802c990a1eeb0b7 : SUCCESS in 8m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c5a69a1ae7648448a9fd9e61b3a02b0 : SUCCESS in 7m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eb75048145934e36800b3bf2c0980e42 : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0174e30267b743ce8d954ae6dd040b7c : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/246b2e8e718b4a929d31231d5f6a1d2f : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38c23c07567047ca80c11382a0a6e11c : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/859f8e34de7345268ff86afdf0ebcb6b : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/497f6d0360a4432685db8d2b7042aeed : SUCCESS in 14m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/edace6431e6948cbad3c4431f34fa14f : SUCCESS in 28m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9b1f14b84fe54990aa2075747d1c621a : SUCCESS in 13m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5de3fb93ff8d4a1c9c4d10a8d1056389 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd517ceac49a4290a12595e313a0d5cf : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e01efe811d92400688327b0ddfe3e1f7 : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af9f255b847348fe8a50041af8e5ef9e : SUCCESS in 7m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/77f9a8d0b96b42d1aaf170ab0fa98246 : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f58a9e9306b445fd8b7bf8467c0c51c5 : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b3ebb55f6a04f15944f1c1ae05ff248 : SUCCESS in 1h 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8f944c30ccfd8960a80a01ddc887e1d0f40c6fc5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 05:53:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54ebec7150b53a74ae07f6700f607da7ee75ec69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 05:54:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8a964bb6627b2a5cc24fbecc8e5cc641a28d152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 10:11:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317ead1c2bc94bccb5dc13fe9839e3a9 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3be04e3983f842bfb0783319a4d53b10 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10f9aae1abac4e0fb14132c36e72c381 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d1edc5ec2374385bfc3b126a16784b7 : SUCCESS in 4m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82117e82e7a641078c538246bf31effe : SUCCESS in 3m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3177fbe52ffd42fcb1c522c92cc21bfb : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9ada5c0999e047f2a4af1596c47f6a80 : SUCCESS in 7m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a42655c6fe746188a2a4dd67a369bd8 : SUCCESS in 7m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a91a195aeec24ffcbc3ce4e85594e4b9 : SUCCESS in 11m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/98ac9657018d455dbb7e25e59880ba07 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e17e37c260bd47e58909195c0d162484 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/67b837990318409fa42cb841e94ca3b8 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/406f4e5574384e6f987b2781fc3b0b7e : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b3ddf9dd22fe4fd3922ad90ccedb5252 : SUCCESS in 16m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/932197bf38a04c84aadcc3e3d7f843d8 : SUCCESS in 27m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/08fdd1ad7e2e4164a8749fb604115dd5 : SUCCESS in 13m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/007c40299e9444eb8c1c657e6b14d895 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/479ab010c75a4ada94e8f54e952e2de6 : SUCCESS in 21m 56s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0a0a1e5303284965b1af7924cd86128e : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/594f882a61d94f54b8a998b6072dab5b : SUCCESS in 6m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/75ae7192ee604cb19258801aadf89efb : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/629f517ecc1e4102a30a7b8513a2762f : SUCCESS in 1h 35m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f99e9b48a22440aad4d563f9d7727f3 : SUCCESS in 1h 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9d8cef69accd2d934e8e605b18d4c46000b60a7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 10:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9666b174f13ae84e9916e2a0f8868974c2d1a083","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 10:12:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5899f305d594fafb7e3b0f919e0bba3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"659488e01d25f5969f16a354c8897c5e417b1b02","revisions":{"659488e01d25f5969f16a354c8897c5e417b1b02":{"kind":"REWORK","_number":1,"created":"2020-12-02 10:43:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/765102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/765102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/765102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/765102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/765102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/765102/1"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:43:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 10:43:06.000000000","tz":0},"subject":"update constraint for python-masakariclient to new release 6.2.0","message":"update constraint for python-masakariclient to new release 6.2.0\n\nmeta: version: 6.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I125f0286d21481ea071454ba3fff85ec13ba0451\nmeta: release:Code-Review+1: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8fb6d76d327fe39101f38833f8653c0280af2a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/659488e01d25f5969f16a354c8897c5e417b1b02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/659488e01d25f5969f16a354c8897c5e417b1b02"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
