)]}'
{"id":"openstack%2Frequirements~765189","triplet_id":"openstack%2Frequirements~master~Ib7e0568c5ea454c014d836b694db2345635d45eb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib7e0568c5ea454c014d836b694db2345635d45eb","subject":"update constraint for os-win to new release 5.3.0","status":"MERGED","created":"2020-12-02 18:19:47.000000000","updated":"2020-12-03 09:26:16.000000000","submitted":"2020-12-03 09:24:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765189-new-release","meta_rev_id":"f5675e7d5a8a7e55b6c80f0f8b07ae3825018e58","_number":765189,"virtual_id_number":765189,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 09:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-03 05:52:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-02 20:24:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-03 05:52:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 18:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-02 20:24:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-03 02:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 05:52:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"53bf9ea11fb3ddcba2b954ad4ce23ac97bb47c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 18:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33d87c1157dc85018a2089fcb6b1f4e84548e4d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 18:50:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a95f8df148df43bc8b0399e72ed7b9f1 : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fb7c884e50166bb84ddf2e99ce11294138dd1870","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-02 20:24:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37fbafd423b96a3cfb6cc7b28aee63914fa982d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 02:01:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3566e6f38c9a44f9ba21a63e24f2da53 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dafa202bfe346b1a3fe34663b109db3 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e8eea1463b746e5930c4f19901b72e6 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3780f2c45fb14509a6ddcf360251cc44 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbf0f6a358634635b884770c477d8ba1 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f514a94662994872a8486d806992ae41 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b704ddba58041cf8a74ac38cff10ed3 : SUCCESS in 6m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d67b75639943412da60f29482e4c56cf : SUCCESS in 8m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee7e5d72cc6640279dcebc00a6c02f7f : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/16c00c8aa8f6418e9b4b91a24eb65120 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/984c6987c6244588be36f6b2bb6146dc : SUCCESS in 5m 08s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6fd2ac53d2a442beb9f7a907d621919b : SUCCESS in 7m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b05d4f98b904342a87bb97c168f3e91 : SUCCESS in 6m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74226e147a364edd88179f4d8c8f4387 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6dce2696998c4df7a443183479b2bef3 : SUCCESS in 15m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c97d30ae965442e38f4da0dbf21839f4 : SUCCESS in 26m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2ace549d2fa423fafca1930c3ac4b8b : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a97f3f6ae8dd4a1e9cc688282954f049 : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/91f924ffd1094c00ae1c378ea48d6cf9 : SUCCESS in 18m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4aeceb02cf3f4e03841d4262dbc59871 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a31de4a5225491896c9ceb488d9985f : SUCCESS in 6m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c81a792cf8cb4dbab9084900f93d0143 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e37a8d6d96e540c499dfb1d0541b0885 : SUCCESS in 1h 50m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b24633b43a74bb48decdfdf9360f556 : SUCCESS in 1h 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"26e349d877f1f9c0bb948fbfd0923d449e4b38b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 05:52:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9bfa3f85495239dcb6801190a757e1718495e0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 05:53:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1faa03e063125650fa4ad8e8f153dc15466dd39a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:24:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a67f4e839f14e01a08876d5f92abb09 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e98489391fb425282af1e5cedea314d : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7950df42024902b3dc03065f5894ea : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63fe90097ec5451090b35972a84f8c54 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5621c88cd8d4940b4c71ca0f35a0161 : SUCCESS in 4m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fc81e5657824ea0803f34e89a7ca0c7 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2214e6c50845454ea38e1bd95920eded : SUCCESS in 9m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85125ba2ef094a3593564c5a03d876ba : SUCCESS in 7m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9fe8526b01854dfeafb38ef242b7c80b : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bde616c21a3946af920afe1a1bacc49c : SUCCESS in 5m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2522b8538bff4d6fa55fdc8bbee7daad : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/585c19e037d34471bb00737dd95d56c9 : SUCCESS in 6m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97b37840988840acade743affe41eb15 : SUCCESS in 6m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4692b4313b2247588d33c45c81795ffa : SUCCESS in 15m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d192c3542c2b498a93d64939c4125f53 : SUCCESS in 26m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/31e49f96340c48e1b46dd705948653c1 : SUCCESS in 10m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/826f52a0152243578273ff17b3523704 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fced79a19f984a32b8f131a9d7f53fba : SUCCESS in 16m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0089e3c6d6d74d3e91b350ac84e9ee18 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3858e9429ef40a9931553899482bcc2 : SUCCESS in 6m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b95f01e42423400eac9c4e845ac9be19 : SUCCESS in 9m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e64cdf6021944e03be6f289bc67df4bf : SUCCESS in 55m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/736649598c1d4e6c82fce94dbf7e654d : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7fcc8f02c3f362b551635a5b7c37b5b75ec3488f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:24:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f5675e7d5a8a7e55b6c80f0f8b07ae3825018e58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:26:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcfedd32aaa4d63be92149e83fff9a7 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32bfd7c8da5c582e5eab114d043fb33c4762e535","revisions":{"32bfd7c8da5c582e5eab114d043fb33c4762e535":{"kind":"REWORK","_number":1,"created":"2020-12-02 18:19:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/765189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/765189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/765189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/765189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/765189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/765189/1"}}},"commit":{"parents":[{"commit":"6dec47e08a0ee95ab787cd4779cba8a1f25713cf","subject":"Merge \"update constraint for python-magnumclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6dec47e08a0ee95ab787cd4779cba8a1f25713cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 18:19:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 18:19:45.000000000","tz":0},"subject":"update constraint for os-win to new release 5.3.0","message":"update constraint for os-win to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I9a6d8bbf4972694e29cff65be684cdd3ce057af9\nmeta: release:Code-Review+1: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib7e0568c5ea454c014d836b694db2345635d45eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32bfd7c8da5c582e5eab114d043fb33c4762e535"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/32bfd7c8da5c582e5eab114d043fb33c4762e535"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
