)]}'
{"id":"openstack%2Frequirements~765207","triplet_id":"openstack%2Frequirements~master~Id7e9f5b04a58c323bce22463a465a4001d32e24d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id7e9f5b04a58c323bce22463a465a4001d32e24d","subject":"update constraint for openstackdocstheme to new release 2.2.7","status":"MERGED","created":"2020-12-02 21:02:09.000000000","updated":"2020-12-03 09:43:42.000000000","submitted":"2020-12-03 09:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765207-new-release","meta_rev_id":"6018e877029d04b717381829c5ec5161bc795753","_number":765207,"virtual_id_number":765207,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-03 09:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-03 05:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-03 05:52:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-02 21:24:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-03 03:21:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 05:52:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a56a099488755a312849fed3b23d43e8c3df3f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-02 21:02:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c07e490804900d48584130b6a7cbd773d137a762","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 21:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/228570f9caf549d6a1b93b4e14c50a20 : SUCCESS in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1cc7657484422c9b036eb3d96443a0e30dc2adf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 03:21:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f581dc97d9244897a7b5780ca87afe98 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2601c43f6e3744a3adebf9dd3615ecfd : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c6097f287844ba87221fb6b32970e9 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d59a2214b434e02bee15dbb4cb72d79 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32d5571b68484bb096a6a4c708a654da : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66852aa61eda423aac9e58272c131a3c : SUCCESS in 6m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cee45d2a191e40679209325cc77faab5 : SUCCESS in 6m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db3637331eef46cf89fb36c5061389a5 : SUCCESS in 7m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a38258fce4346a091930054b9a2e96c : SUCCESS in 9m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/410af9d612cb4090928e38427b99fb00 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/705ce64ba9ee46f4850ee8a15a2ecc81 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e2635e6b553f4058950aa22565bb79b8 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a11a7d8ec11491383cd7925ac4349ed : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2dd55831e16d4a14911f88280416186c : SUCCESS in 6m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1f33334cddcd4f6da6bf8354f96ca639 : SUCCESS in 12m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/57cc4deaf7b94bdd954c1879cbe71b9c : SUCCESS in 29m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/af17859aa67841418720127f6797b73f : SUCCESS in 12m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ad75600d7f12402a9a3be2a93c1b0709 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6f050e9e5dd4f53870e2255ff727ae4 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1b2eecac76b64f76bc89c13c5eb23b3b : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e56ea6758364798be2410cd07a5a516 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/632032cc101341be83fd6122bbbe0285 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c0927d3e0f34451b5df86718550ba52 : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dee593cf78cc41fd9a102fe359e83d56 : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4a8dc035acad3acb2691ad9572fda70542edcbdd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-03 05:52:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ca730b3b4cb3945a660c24047a76c26e58ebed4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 05:53:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d38cca92c37b732c99fd818180d1f5657bc1c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:42:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11242b451e4b43d085ab0d707f926d03 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfcc3c0609149ad87d34d78d66c41da : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7d24b97e26d473993565d8c6affbe1c : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1971d0194332431c92bd2800c456943d : SUCCESS in 3m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/375c77ea2d5e43c88235ca75f8fe7f7e : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9e5ceb0af71f495c81ee7c55dcac3db5 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc11b7d47c2748989263d3add32943a5 : SUCCESS in 10m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a690274e2e749ccb2396fe77274dee4 : SUCCESS in 7m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b8104314b3f34737b6a728a52646f71b : SUCCESS in 8m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e2908f61ed5d42a28332c7468496635a : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc356951707647fcb5b855e5bcd73485 : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3458dbb908e4ecfb0a69802a13e34f0 : SUCCESS in 6m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13c7eb7d91db4a199e8565eaa38d28f7 : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44ec5b7d2ec84368b9324ddd52fc465e : SUCCESS in 13m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d350076b677543e28a07096a44cc62ed : SUCCESS in 30m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e9cfa6b8da94c618f7f2a105033017a : SUCCESS in 11m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/91f4748448ce43f3874f4e8e0ef028cb : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d8de8d25fec4e4ab391061756205a6b : SUCCESS in 22m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4beff1571b48482fa8132242d5e15edc : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f66ca021fde14378a13e1c477168c446 : SUCCESS in 10m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6e67908ae8f443bebf7aa876d6a6fd47 : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85538b9e499646768fe3a953d2fa7ea7 : SUCCESS in 1h 25m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75db3573ddcd460b87193bab7417a747 : SUCCESS in 1h 25m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b2fa9e9842ef16f9b1f991d3ce6d3f81e6656776","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6018e877029d04b717381829c5ec5161bc795753","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 09:43:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4669ca27ad2a4f50ad57358e5063414d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7050412988306ed395051718b5cadf884ae4323","revisions":{"b7050412988306ed395051718b5cadf884ae4323":{"kind":"REWORK","_number":1,"created":"2020-12-02 21:02:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/765207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/765207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/765207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/765207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/765207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/765207/1"}}},"commit":{"parents":[{"commit":"540ded8231a0a21343264824cf7607c8bb978a5f","subject":"Merge \"update constraint for keystonemiddleware to new release 9.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/540ded8231a0a21343264824cf7607c8bb978a5f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 21:02:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-02 21:02:06.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 2.2.7","message":"update constraint for openstackdocstheme to new release 2.2.7\n\nmeta: version: 2.2.7\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: Ib9affcca4f4ac08118e3de11c52210dd89731966\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Id7e9f5b04a58c323bce22463a465a4001d32e24d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7050412988306ed395051718b5cadf884ae4323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7050412988306ed395051718b5cadf884ae4323"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
