)]}'
{"id":"openstack%2Frequirements~765346","triplet_id":"openstack%2Frequirements~master~Iaad09f9aa9889359a4c40c4d7003dc922ee252e0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iaad09f9aa9889359a4c40c4d7003dc922ee252e0","subject":"update constraint for python-manilaclient to new release 2.4.0","status":"MERGED","created":"2020-12-03 16:03:37.000000000","updated":"2020-12-04 06:34:58.000000000","submitted":"2020-12-04 06:33:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765346-new-release","meta_rev_id":"bff362dbaeaa090bfa540e3f68ac2eda7b7e3a2b","_number":765346,"virtual_id_number":765346,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 06:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 04:18:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-03 16:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 04:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-03 16:16:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-03 16:33:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 01:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 04:18:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0d3b19edf17cbedc4df9210b9c8f16b10c3fb0ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-03 16:03:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de4a8074d0ce7bee79c125347616a4d522328f86","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-03 16:16:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f1b139cfc6a1a30a96eeafc11f5e102968589b9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-03 16:33:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d877ea8e109c434291e3146cfc357e03 : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"92b7ac7691972ac68530bb506d049c4b029edcaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 01:44:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd2dc4c8d6f424cae0ef3ef40a912bc : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007cd4fa09a245a180c9c4ee16835a30 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e137a765c24e41fdb96a932ec62e4737 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/754974a031b949659864593e646c3f4e : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af26064691ea40b78156d9277f70ab1d : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/844b440609fd44e5a72346591b9493b8 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d3434d84c8e4732a6b999c3c119392d : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/15fca1c91e08424fbb610df04eaaf24b : SUCCESS in 6m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d3362a5d2fe84e7fac59415c3c03c2a3 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7ea7b1a8be1a48148ff169ec5418f890 : SUCCESS in 13m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf26e43be33c40b3b2dc787a8a4e0fee : SUCCESS in 6m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fd18a2ac9d8a4c0b95ff0fa60222f054 : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/671550a11bc24bf2b62668e34a900e2e : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85050420706a420ca50cdbda5d94da2b : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a469d5923253477cb0105922b4c77b85 : SUCCESS in 14m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/138f583501d748c4a74bb4d09176b4e0 : SUCCESS in 29m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7785bc0da7684641897375cf30bc7c06 : SUCCESS in 13m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d09ff2c6ce347688686e04ffdf94b12 : SUCCESS in 3m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f54562afcfb4839ab728fb71d92a00f : SUCCESS in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/24b0287ddc244b3c8c586f6d9ba21ec4 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9c230f48571c4b0291ddd62c4d7483bc : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5584e7dd5ca44611952431380060c102 : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fa823a414c24ba6a0e25d310ec2047c : SUCCESS in 1h 06m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b458f632f6b04853917cd9d4a13ae30a : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3ae08bfa87b02fba3fd1cd1215a64a43e2d20e93","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 04:18:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1ffc4c6cec71b9bcfba3b5455c1d3ba709e8e289","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 04:19:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af3f8de96041bc26c73b4ddcbc45457fbb82ef3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:33:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be575fc3398b4b4a8af9f1bc31963092 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59efb45b0ede498f9e14cbdf1a68b729 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a287044dbc404342b0206dcbabeb3f2b : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54b79357e15f4d87904822d091e5cada : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed30d42700e84ad0be346deb1095b6c1 : SUCCESS in 5m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00fe747d80d44192853f89459691db01 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a779d67ff0d4b4ab7e8a3a523618046 : SUCCESS in 9m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2b112169a1c94339bbbec36cea92f8f1 : SUCCESS in 10m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/017ddfc14ad340aa8fb8c68d93dd73cb : SUCCESS in 16m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/823e2fef6e6c49b1bcdd284ebf5b433e : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c908f756b38e4eba9c6c4d50af627ce7 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83ccc4d8ed2d43959704570a99466ac1 : SUCCESS in 11m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e1652b43e91143f384a3a623da42423e : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cbc457a1f20f4004b87c97895431846b : SUCCESS in 15m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/864472101b024040865563fcd555622b : SUCCESS in 29m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/75cbea017c314ddb97b4ede1296c6e95 : SUCCESS in 16m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cc1f83b4c43245368251a1c3cebfc853 : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/58274e82beb5489281a5dd8f3480ef9a : SUCCESS in 17m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/62a3858b39ba4240aa0439f24b40c01a : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efd5bf4be6114ded8cccf12cdcc2ee92 : SUCCESS in 9m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1786a54b1b6f4a848870cb6706c8d7bb : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc1af963771d46198396d6d1108115c1 : SUCCESS in 1h 03m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34a9e0c6672d4ce1a288b5d09672022f : SUCCESS in 1h 56m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3a9eb00da2e69f4ca4db114aeed62fbc250c7182","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:33:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bff362dbaeaa090bfa540e3f68ac2eda7b7e3a2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 06:34:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f47352e57f479aae7d85b484cd9714 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f26dffab0b9a30fed74bb18f34e5376885581112","revisions":{"f26dffab0b9a30fed74bb18f34e5376885581112":{"kind":"REWORK","_number":1,"created":"2020-12-03 16:03:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/765346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/46/765346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/765346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/765346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/46/765346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/46/765346/1"}}},"commit":{"parents":[{"commit":"de0ee4fe7d589bc0f3fd175109da76a01c57c2a6","subject":"Merge \"update constraint for mistral-lib to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de0ee4fe7d589bc0f3fd175109da76a01c57c2a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-03 16:03:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-03 16:03:35.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.4.0","message":"update constraint for python-manilaclient to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If28866e25221db8dbc779e344d107d9924179aab\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Iaad09f9aa9889359a4c40c4d7003dc922ee252e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f26dffab0b9a30fed74bb18f34e5376885581112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f26dffab0b9a30fed74bb18f34e5376885581112"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
