)]}'
{"id":"openstack%2Frequirements~765496","triplet_id":"openstack%2Frequirements~master~I5fbacaf17be37946c59181a6ecfd1b40101e887d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5fbacaf17be37946c59181a6ecfd1b40101e887d","subject":"update constraint for oslo.config to new release 8.4.0","status":"MERGED","created":"2020-12-04 08:47:30.000000000","updated":"2020-12-04 18:24:19.000000000","submitted":"2020-12-04 18:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765496-new-release","meta_rev_id":"ff8c36e282708ed5aed7706c4653c9d86f10ed91","_number":765496,"virtual_id_number":765496,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 18:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 15:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 15:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 09:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 14:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 15:35:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ea47365df63c56823724c68720f0dde0ef2d09c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 08:47:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b4ef4d4788b30f47413645a77c3e076ce1339e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:16:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4c1659cd98414eadad197ef2f30467c1 : SUCCESS in 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d41019664aaf16945355825d32fb726351937c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 14:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4b97db9573450e9307e4cdf1b63faf : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b68ce1dcfaaf413fac2d8145a49116ef : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dd3107dbd234bea9cfb547aa76f9c05 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b710e2e421ba4f18954ec7faa00bbeee : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92bb594dd5044949a7d3fb65de3df26 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/02cf969dce1a4342b3fb34e233a122c2 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d1a890c188a4f71aa0dd21f7e879f56 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c7de623868a84f87ad611b405e0f56e4 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e31ef3d8e4e840a8ae66f4f3f3026c4b : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8e92eda35ddd477781fc207705d67046 : SUCCESS in 11m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f18250fe7b674ba3bb37fa37275a3419 : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8811c79bc3c6464fb8294c3a401c2365 : SUCCESS in 10m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a131c4ee6d44b6a9cdf89359df24f0a : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9b99c924778c4e35b2ac76f9c762c203 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e90d36bfc5d40e48fdaef8c9f656220 : SUCCESS in 14m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/13e0865b9fe741bba4556a673cc09a57 : SUCCESS in 34m 59s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/780d7aacde534033bc219da172a99660 : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21c984580d674d6c85e0a0f66cc8e39e : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0755afc890654c699271a523e402a120 : SUCCESS in 17m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/166eaf4fa643472695b3bcb543098ddb : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f667e75d3dc249589d6345353852333a : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13566dccacb149a28b65d4298d074a7e : SUCCESS in 13m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7067079f1b22403ca90db2b3f4c8ef72 : SUCCESS in 57m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/323803c142c74d56822cceb6b88590a6 : SUCCESS in 1h 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"461fab267f3074698ee1b36d6e0c7eaaaa5ab5b3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 15:35:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0f662864a248b09a2955b50db0d19d06cec7a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:35:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53c1957893b6c7fc7aea497bcf77772d013283b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:22:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4943c1a598774a6abe729c21b4fb3c39 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e72e0dfb074173aced1d8b6e8adf34 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54b3a3dd0ab9434a845d820c9f9559d6 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b97c350039c54ad980cfd04eda00c211 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/283cae078706469f8359febf37a35fcf : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c10439fdd3184ebc86119d775a1c24a5 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f68e0eb3b58465c99ae8851149f987b : SUCCESS in 9m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/501656e99b3749918a9f60e0d0e9f372 : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/48e6e04daad444098da243088643cc92 : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18b8537f498f404c851a7309bd412418 : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfdd27e82948489c8df863e54df5e7a5 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92a8c7a3c16843ee8bd01a729ba59f7c : SUCCESS in 7m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/66b721a8658e460dafc73e3fe8109e64 : SUCCESS in 9m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7480a49c6ba34f44a95b9dd981b73f89 : SUCCESS in 17m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/44f380d441c14c8a8aa4e828d16b5932 : SUCCESS in 32m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97d875bbf6fc49d191b948a9a55ea00a : SUCCESS in 14m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fd460efd9ae54000a343f83c555a009c : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c6fdafe002c0448bbdf0a2b4f290a6ec : SUCCESS in 22m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d48545fc79dc4d618553a6e4788cf6ff : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddb98ff839294a22836bbbfb1d484d0e : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/518b3d63dd494d1eb3c23ac31bd7b7da : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d71f5d315e4482ea1a999e632600cbb : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8102004bb6134e67860c6825b74385cd : SUCCESS in 1h 17m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ad490f35020c4d4f72700b141dea843fbfb3aa25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff8c36e282708ed5aed7706c4653c9d86f10ed91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:24:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/241b55a76f1a4f04bc2ff03eb6b28603 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d9d84f6cc6becec2ae6f2c61aa1be88295c2c84","revisions":{"4d9d84f6cc6becec2ae6f2c61aa1be88295c2c84":{"kind":"REWORK","_number":1,"created":"2020-12-04 08:47:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/765496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/765496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/765496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/765496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/765496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/765496/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:47:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:47:27.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.4.0","message":"update constraint for oslo.config to new release 8.4.0\n\nmeta: version: 8.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5abd27c55a1f6f83cb0f9a83a4b7169449fb2707\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I5fbacaf17be37946c59181a6ecfd1b40101e887d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d9d84f6cc6becec2ae6f2c61aa1be88295c2c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d9d84f6cc6becec2ae6f2c61aa1be88295c2c84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
