)]}'
{"id":"openstack%2Frequirements~765497","triplet_id":"openstack%2Frequirements~master~Ie741d5d87af3933ff26eaad75fb701608833be36","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie741d5d87af3933ff26eaad75fb701608833be36","subject":"update constraint for oslo.db to new release 8.5.0","status":"MERGED","created":"2020-12-04 08:50:12.000000000","updated":"2021-02-12 04:20:29.000000000","submitted":"2021-02-12 04:16:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765497-new-release","meta_rev_id":"029099d592f474c3f57f2b56bec4b5d08ce1ee7e","_number":765497,"virtual_id_number":765497,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 04:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-02-11 23:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-04 15:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-02-11 23:02:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 09:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 15:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-05 00:24:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-01-04 15:17:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-04 15:35:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-02-05 14:58:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-02-05 14:58:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-02-11 17:22:25.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"}],"messages":[{"id":"9bcb098456c42ff859f8b4100a1c4aa2d149552d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 08:50:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"608241ab1408d7f4e76dfd9a23cf7314c0ab7bbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:17:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bbd87d8e80c94a469450bf4502463e52 : SUCCESS in 22m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c10ed8cab0897bfe0460cd580e320903f8d6f593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:08:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3759b63d7010473aa5d679799f0cd5a5 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca11281a9e04259b2a1aa5202c4a499 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a09acca670ee43719828a44e5fa81d92 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da8edc9cee954bc994d019200473e4a4 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1709ddb826b4ac5b296796a2156d04f : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c48da4a1a0f14341b6ec85c3be9643bc : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a8e09ba2491d46218f0131a246a5c54b : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b0fa6d9158d4d638411cf279a1fb44e : SUCCESS in 8m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/97fd635dcdd24babacce54099239c32c : SUCCESS in 9m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/efa670bd002044dfad364658ccf7cdd5 : SUCCESS in 9m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2d63a84300364ce79af36e839077221c : SUCCESS in 6m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a0a6910f1d84e81853b786bd6035897 : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/804e4b737380413183b788833b1c6228 : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad7755076e7e4b349a32689e48ef16be : SUCCESS in 7m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/40db5d7fde584b41bfc3b0f2cb633028 : FAILURE in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3071d0d3c089424d92c6fc1ae3d72f6c : SUCCESS in 25m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f197f571d11d444199dc2bf172ff0183 : SUCCESS in 12m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f8e9807833444e0aa6f25e5c33265cfb : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55499e5b5c9e40cebad18a94a0c834b5 : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5669a8043d664a7fa1aae8198d8e4827 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/649d42532197468c83727157fe891585 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/368478946a0d4107a401db17f7fb4dfc : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60ea1ce6f42540218a577c73f3d3232a : SUCCESS in 59m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1da525c67a3b4488b5a374eec9e95061 : SUCCESS in 1h 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1987d86029af3fa0a04807eca4d3c9d9d3dcfe30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-05 00:24:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"de5b9c9c3be8d7711899669025707495682ed81c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 00:48:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d86c995660c84e989ea0888498932b45 : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e47cfa25231fa41196b971f71561bd5e6ebf0231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 06:08:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a25d4724af4703a3aa84486039b85d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0c9027dcca43f893f9060788ed04b2 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/571afba4572c44eea78e9c1b2c2776e3 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48af679d17e54918a22b8732f5e80b59 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78f9c1802eba4192bca5022490b56ae9 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55a0e1e9aab9463ebfd9ee2099680d8b : SUCCESS in 6m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc57f381b18e4b558a53a7baec4b7dde : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8d8c69c0b6145c7873d6322e1614928 : SUCCESS in 8m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4edc9a0d6cac4fd5823d69b4f5040ea2 : SUCCESS in 8m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6df53fc239464147b00cab7a888f4f7d : SUCCESS in 12m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ad4529cbb7264496ab6d0fb8f72db25d : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4aa562cae478479d979f31d0b3a52ed0 : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87a22419db514873a2958246338c24b4 : SUCCESS in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8631e4d5f164fbfba9ec025ad938f58 : SUCCESS in 6m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5614c6543b004fc0b1e6e24f68e4514d : FAILURE in 15m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/64bca8853b9c4b67a4b8669a174c71ae : SUCCESS in 27m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a773f730b761477d87d7ea48a1e10188 : SUCCESS in 15m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f66e601f848b48448aa1fcaa3510183f : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e63be250c0bf4066aa44b670de121798 : SUCCESS in 20m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ee10caf8801f4a6bbda2c0a0b2c7334e : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/656098399a6046a7a15d3562cb014719 : SUCCESS in 7m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2d5bcea0334348238601ebaeafa7ad91 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef49d87c4cda4f038269bee43f2a5f49 : SUCCESS in 1h 04m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8a202f631ae437e8919e52b8e46add5 : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ceaa1f65ea1c2ccc237be55fadc78fdb80b69baa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-06 00:54:36.000000000","message":"Patch Set 1:\n\nrecheck\n\nseems like a repeatable fail on keystone-py36","accounts_in_message":[],"_revision_number":1},{"id":"ffd98b48a19925702235c688907a39962e987c7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 01:15:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dd1c84c4ba2048168a7837b767f30191 : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":1},{"id":"30feca1c5c7a3dbfe2ee47e9d60230b521a8f7e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-06 02:44:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ea1f3b2d80e47a798b4e49173b08417 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad04859577454cdbafbcbceff73235a7 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f72a1ebe4324b23897f918391cc7128 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b5f2c200a2c41c98666d65d3a016e25 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32032d46fb5e47c493292882c468e287 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/587f43ac604a4a5db80959d954079bf0 : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e6078a25c0d14c4f8c3d74faf5cb8d9c : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d2f2a6d44e1401390e5affca63cac95 : SUCCESS in 8m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f44a49107a84474c862efd2cab74a8e2 : SUCCESS in 7m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cbb01a7607c84beaaf624b007ed38acf : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b816d6ec38f400c8602c483752e4858 : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/de9d83e22465481383a5d9b11a193ea8 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/213a48571d20413b81dfab91fd6e7355 : SUCCESS in 6m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35d020b6ad344edea44d8283f85d1153 : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2d6e25dab2ba44ef81779152828867b3 : FAILURE in 15m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/141430fbe9c643c697da1ae6f84201f5 : SUCCESS in 27m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6a414d40b7749fc95ba37fb6e2f5c04 : SUCCESS in 15m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/332e272cf62940cb833d5506dff2369b : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1910a7ffa0764a81b023cac7fc459994 : SUCCESS in 17m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dd738674f8b54fef9e656f2c3beac459 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8905c0538364274920d5df49ba6c959 : SUCCESS in 6m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f5b76c6374dc4632970a5c830387deee : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbeb9a25f2bb4866a999d6d870caca26 : SUCCESS in 1h 04m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be223d11bd64472788b6bd1a34ef0427 : SUCCESS in 1h 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"48bd23f9290395a070043c9ac356b04ed6ef8a8c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-04 15:12:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c23825de1f81a182bcc8b4efbcfcdc6667481b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:12:37.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b53c87961bfd5fee82a07ca662f20adaeb6d76c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:12:40.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"0cf4de027323724619d0f7a11c970fd853a7046b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-04 15:17:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e396ca3a1d4df806bda924ac0a1ee48abc2392b1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-04 15:35:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28e324936d14092dc0182ea6ad094cf1e73186aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:36:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3b2179948eba4b218bf1fd2cda3391d0 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":2},{"id":"1e78701f00e9a68b546b694440ecc97b0f37c2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 16:53:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d55fdcd767a749989d7afbab0b02cfe8 : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04c512ee2ba44747ae81805ea0174a99 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac4a42f5cec14941bbc33186b6bdc83b : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8ade3b66f8547489150a7b211173e20 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f6205fe87d4788a27b5e03f699537a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/49d625eb6fbe40f29ed8a58b29066dd9 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eed84a1f09b443abab43382c64f0b210 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/12e834c702b04e36a8d07363ac080077 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c49915241776440db2d0206e5d3e1822 : SUCCESS in 6m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5e8efcc6f3084c0482fa3c0daecab9af : SUCCESS in 9m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/18f901e456b940ba92190aa1e2daa558 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ed57e15ae31749b3b10e1f3f8ce54c12 : SUCCESS in 12m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a391036670f74542be58c4cac38e15ae : SUCCESS in 6m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ff0bb91d624b4a1e9da99081971fa37c : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/93390590a90f4ff59b3dc4d008371c41 : FAILURE in 18m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f80709c19a3478abe48dc7fa8679f26 : SUCCESS in 29m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9f157252c60a4ef7ae3c8051f78ce5ba : SUCCESS in 19m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3b5237e636f458f8c4ed8a12b523f16 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1e0d5177e8845b9adec081fce4878aa : SUCCESS in 17m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7e5f2ac204484e66ab62a568799b3f03 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/70aa7fc6803b44fda9cbd0c6ad345275 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c25d827e200e448dba05dcd3a547fad5 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d727095e1c534cd3b329627ece53fa2f : SUCCESS in 55m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/871f676879a34419bb6b52c10a3e479d : SUCCESS in 1h 22m 41s","accounts_in_message":[],"_revision_number":2},{"id":"824038be0d8489ca68570b65f0cf801f95388a0a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-17 20:53:11.000000000","message":"Patch Set 2:\n\nrecheck\n\none more time before I bug keystone people again","accounts_in_message":[],"_revision_number":2},{"id":"5d77cb687094b8365fd22353de8f49eabeff222f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 21:10:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b2e0c8c640b4697b87ef2616797d70b : SUCCESS in 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"1dfb1d11fec6d5b3a12b548bcf9b33aa8ac436fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-17 22:15:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1d21bc4211458a8194d44e062ed4d0 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8aa138808db4564bdd5ed756c1b9a13 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7654bf755ad9466b9eae6e4f6aaf7b68 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1898f52eea1b49378c707f4ac41b1ed8 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccf8644e5c9c4796bf62bf6ea3904314 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70f27e3e90e74ce4894164a2a189a2f0 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35756297a6124caba54a4fa5af879b6c : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e785f2830864fd3b67b478dc227f5c5 : SUCCESS in 9m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/916238019a184c559fe9d729ef14ac71 : SUCCESS in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/46087aea515d46a68de79b54ffc30e3f : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d9fded43f57f44f69db6fb43e2196ddc : SUCCESS in 4m 50s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a3a3ccc6b8e4743a9f04009669a2e95 : SUCCESS in 7m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e5cfecadb1534432b4792c26c1447d55 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/083abb28b3614144aba9780a568ec4e8 : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2553f056cbee4ebebd21d57217c9ee10 : FAILURE in 11m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bdab958b3cac4efd8dc3733c55c52124 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5bca6c8fd60f4b6ab1eb062c5377059a : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/90ae3f6f19a7469ca7fb2d74f562d9a2 : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8e1943ccd9b45ff97ae0df91544291f : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d381089612f94e8fad2b96a3ec337ca0 : SUCCESS in 4m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c066d89803b44487b55bacb3ac479aef : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/abbd96a76054468c9f6cebfa2e1f2ef9 : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/486f4feab6a1465f8b136b7d8a649627 : SUCCESS in 50m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b6a6bbab7d14616a01668e923457dbe : SUCCESS in 1h 18m 30s","accounts_in_message":[],"_revision_number":2},{"id":"22dc1647ad612d25e483826643cc458205ac3907","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-05 14:58:15.000000000","message":"Patch Set 2:\n\nKristi, looks like this is causing failures with keystone DB related tests. Can you take a look and see if the failures are legitimate?","accounts_in_message":[],"_revision_number":2},{"id":"234cc04fe5797b63147d963817a53c99ef483087","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 17:22:25.000000000","message":"Patch Set 2:\n\nrecheck \n\nKeystone merged a patch to ignore the warnings until we can find time to move off sqlalchemy-migrate [0].\n\n[0] https://review.opendev.org/c/openstack/keystone/+/760678","accounts_in_message":[],"_revision_number":2},{"id":"fa738b4762a8a09dcf892dab7247bc2fbfb2179a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:53:41.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/85d51880949f499fa99ec512c4e7d1cf : FAILURE in 27m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c7467a7aea6c84fe6246bed3882c96016f6f6ab2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:53:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1004d90590a74bc78e2a46634de85771 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f39912fb6ed46ce8e0bcc695d0d7946 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56452b4491f1457b8d6a144c5a2e8a5e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8cf4bc84dfe42c19d4fc222837f990c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bc28749306e442c8bddcd4e4fd7fe97 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9c92fb83f0dd4e6f8d9adabe0ad02fba : SUCCESS in 8m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3da1228fe23b4b50ae55b52a7eacc882 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10275e2baf9344cd82a5cb939df8bf04 : SUCCESS in 12m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c211238f85e45ce9f88bfddcd4b0e07 : SUCCESS in 12m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9665042131364f848afaec5a774a065a : FAILURE in 15m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d00517f181894f8f95045846a850a598 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d1c23e6b505e4648a407ef0cbc8d8f3c : SUCCESS in 7m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c005fd3b8484a38bdea913e547af3ad : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1159a07f712f41d4b87a42b0d9ce11e3 : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/66231db811f344729777040b77fee6c7 : SUCCESS in 20m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5cebf31676864e948fa9d17e8e6894f8 : SUCCESS in 27m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8eb3d942cd0d441796eb3e3f282b7e94 : SUCCESS in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2490d1d5c5214cbeadb3e49bd05ac9a5 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0b82d4d021414d8599ff87c600b84659 : SUCCESS in 17m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d9ce3b1ea6614998ade008a92262c941 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d07a2aebb60c4a3f8d579a293d28aa1a : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0a6beadfab1b43ef9422ea4408ef9df7 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83c2646275d842f8b17d73cf2282ab5a : SUCCESS in 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/295e67bdfac743ae96d0378aaa35b0c2 : SUCCESS in 1h 40m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a9800f54fbb0dd3c66841ebf74e052b6d37956da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-02-11 23:02:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"245ba528c6149c413122cafc02bfae58529fc23f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 01:28:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bde539b8a0b948c3ac9a3ccf1e5ed6e0 : FAILURE in 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a9414171fc6f25124c39469d5ead1d5fb8323cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 02:08:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0174547ac774d48b9d8880b8a7d4089 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10155aeaf924ab09a902f587891d353 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37c3d7633c164238ae0b860734dddb83 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8da280c163964e429398f0c50ed63738 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6b61627edfa405abc404bcee82bfcb1 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6abf5432c72b405693e78460efab29ed : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/56afdc793081459893d51a8cc66ba6d3 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6e10f1beabf6456185261dfb27e231da : SUCCESS in 12m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c4225cd853447c59682a5a143d279c3 : SUCCESS in 11m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9581a8f2eb784b588c52e464b184d454 : SUCCESS in 13m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a97c37a95e1143a781ed3bd2ec72b50b : SUCCESS in 7m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6f398bbe6b6c4457b0d97e66d939a2c5 : SUCCESS in 7m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3833e447380c49ac83208cfca431f168 : SUCCESS in 7m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d892ef7571744221bd9fc0410584bda1 : SUCCESS in 8m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/58e9723118134d2285069c0c4ee6c69f : SUCCESS in 14m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/640bd2835b234e4a8d2d3c256b99ea3a : SUCCESS in 31m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/134e3ccc57544d4da1a9a68efdaf4d4a : SUCCESS in 9m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2574647291b0495cadc0e5d0c31bd299 : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f44cafdcdec472ea23bc88f38158ec8 : SUCCESS in 18m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6d134093044944d7906fba5dfbcdcc59 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9bbe2fc4a1dc4c7ab1047b3c635b0d6f : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e754e766bda94a62a87d1ba083b1451f : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cacf2024e1d14eca8265547d84d554d8 : SUCCESS in 1h 01m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b9d6ed76ae304b9aadb0a488f1c9e1e3 : SUCCESS in 1h 28m 28s","accounts_in_message":[],"_revision_number":2},{"id":"f7f5bb671526dcad596c3a6f227b6dad1a6cd313","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 02:09:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"38f7d915e9d8e3382f9cf30b176f99f818aa7781","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 04:16:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b656d535f9a4cf3ba1c5b4742533a10 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6697fa16bba04602afc1d78a5744f49a : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f03eac4716fc42e4892959d8bec4fede : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a046070565740f5b6b594568edaac99 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f378159616924453849f6afc8a7b92cb : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/926ea9ef704b4e7ab20f09adab7e07da : SUCCESS in 4m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/61defad29f7e4d9eb14a86b2096f88d0 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee3e96c48f854230b0dd146663443969 : SUCCESS in 9m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/43dc267b64ba451298e237425b9347f3 : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1b4cdd1bbe674748ad942c768c2e56c0 : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8209e8ac7b1742f49f6fb95498a91174 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e01ecb1dc3fa4bcaae3d428b5de40fd7 : SUCCESS in 6m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/746bd97193ff4d858e06864aef496575 : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6fa86bdfb4494eaaaa383bee4b489386 : SUCCESS in 13m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/551638234a4b40c49fa28217b03ac9e8 : SUCCESS in 26m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2fae584160a849bfbcee72bb80b5dab6 : SUCCESS in 12m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1927235f89af4a088b8b79234ed1e152 : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90225f6f1f224a4894619fd5d0c3cd53 : SUCCESS in 18m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f60a90a8e5842fcb1b9bf09677f1200 : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4231a72b23b410a99f9515a9a2b5736 : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/16a0cac5978641928ef8a2d3cc04c83a : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4633acbde71c477b8c3cb4f70389b57b : SUCCESS in 57m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edbea92a9b75481bbafa15377e167f28 : SUCCESS in 1h 23m 37s","accounts_in_message":[],"_revision_number":2},{"id":"dcb296c83fbe22f921328528bc98a9a38dabb6ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 04:16:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"029099d592f474c3f57f2b56bec4b5d08ce1ee7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 04:20:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecc6659314044bcb6e31928acc70843 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79ef5b66e49f05c59a0f71bb4ffc1db9f227256c","revisions":{"b1940ca10cd3d6027b1811dc1dfebde9dad6c8e7":{"kind":"REWORK","_number":1,"created":"2020-12-04 08:50:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/765497/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/765497/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/765497/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:50:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:50:07.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.5.0","message":"update constraint for oslo.db to new release 8.5.0\n\nmeta: version: 8.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I2fdc45b99250ecb76f61b96cb80293b220b62b5b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie741d5d87af3933ff26eaad75fb701608833be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1940ca10cd3d6027b1811dc1dfebde9dad6c8e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1940ca10cd3d6027b1811dc1dfebde9dad6c8e7"}]},"branch":"refs/heads/master"},"79ef5b66e49f05c59a0f71bb4ffc1db9f227256c":{"kind":"REWORK","_number":2,"created":"2021-01-04 15:17:10.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/97/765497/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/765497/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/765497/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/765497/2"}}},"commit":{"parents":[{"commit":"8d035c52a99666ee7a56d79e139a8bc00da09308","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d035c52a99666ee7a56d79e139a8bc00da09308"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:50:07.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-04 15:16:57.000000000","tz":-360},"subject":"update constraint for oslo.db to new release 8.5.0","message":"update constraint for oslo.db to new release 8.5.0\n\nmeta: version: 8.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I2fdc45b99250ecb76f61b96cb80293b220b62b5b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ie741d5d87af3933ff26eaad75fb701608833be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ef5b66e49f05c59a0f71bb4ffc1db9f227256c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/79ef5b66e49f05c59a0f71bb4ffc1db9f227256c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
