)]}'
{"id":"openstack%2Frequirements~765498","triplet_id":"openstack%2Frequirements~master~I58ca6b0484ffd530e40904abf3372cc8483bfb95","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I58ca6b0484ffd530e40904abf3372cc8483bfb95","subject":"update constraint for python-heatclient to new release 2.3.0","status":"MERGED","created":"2020-12-04 08:53:15.000000000","updated":"2020-12-04 19:04:45.000000000","submitted":"2020-12-04 19:02:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765498-new-release","meta_rev_id":"a1aa9284850e02298b852019ca4f00bec6982f55","_number":765498,"virtual_id_number":765498,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 19:02:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 15:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 15:35:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 09:21:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 15:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 15:35:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5849bcf6d70705ac115632402e3aca8b5b3ed0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 08:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66b99e4aa74cbe01c04e26d8b00489ec0137657c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:21:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3c6e73c429d14f328ac35f6b53ff65ac : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1baeb5585eb20fa4e8d5b9eb9c6b0f9136370b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66aef6ee27804cbdafdff69be56d9197 : SUCCESS in 9m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12eb5642b6894156a9ba0fe7e0ee06d8 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1957bc48bb914b3e883828e62fbe875b : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f03ea52820432fae0994b3d7a79f44 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a5e8623f14f4e07bc37af0c7a1e6dd6 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e1eb5c49de334aa9b3184bddb197a052 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/353c4b8fbd464ecfab689864d854076e : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4cc51558b7994858a5e21d32ca3db62b : SUCCESS in 8m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27573fa1550e4e6790960cc5279aa4b3 : SUCCESS in 10m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c921d81c605342fa98b3867be66af9e9 : SUCCESS in 10m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aeb90c1aea49459e98d6bfab38413a92 : SUCCESS in 5m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6735d3a413734df9ba21ae7ed65984c9 : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84821f4c541c4ce99bb0e974efad2cbc : SUCCESS in 8m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/37c55b881f614d7a8b62f88abcff5d67 : SUCCESS in 6m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1a75edb5432e494aa829f8dfe1445a1e : SUCCESS in 27m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/04979b0be4d94f0ba9ac60ebc1bac4a6 : SUCCESS in 28m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7472ea1731074727a9f02383a66d4da4 : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/343cff98dd7a489c918c5859316fa696 : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75d5620a993b47dcbe1b79c19d2a6ee4 : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/83d7161e80a348d4ac48d88fb172d8b2 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/167e14d779384da08a47960474840643 : SUCCESS in 8m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6b97b1d9bd2a4ace9fbd23dbbc37d306 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b304fd066b940f4ac3f0761576a94c8 : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c2d712adaad4dd69badedf543a78a76 : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"aeada70c5dcd72ee4a2be2464b0f553812c86284","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 15:35:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4416731464a608c5d8527e872ff47b58e42b2456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:35:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b58fb31f8f620f1aa9d309879d23bf669881eca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29ee2c1714d54acd8de131088af52043 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179ea06373a4470a9315031e8893fb74 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d5d17df99564ed9b70f85e9a845e505 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f36e4399ca964cbba43f350439ff3b33 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e0f220ff6de4a2bbef8da54be2b8c9f : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/86e459d4593e41dc887f8b3377511582 : SUCCESS in 3m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9f73f396f0cb44f19aa6efd0886e691f : SUCCESS in 8m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b7b62807e8a461191267e9af1ea60e9 : SUCCESS in 17m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6e1859646e06436baf8eca75665017ab : SUCCESS in 13m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b82d5b3f3b8e4bb68c5a5491f2ec8bbc : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f208ad0150c4719a93151977edd75c1 : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bf99481d19764757ae14d7e603da8a25 : SUCCESS in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eddb07b3cbe245e4a2fc8fec4a07ce73 : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f0ff7e8c295d43a6895ea25f2006d58e : SUCCESS in 18m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5e3818c5c1f464d95c48feef89266db : SUCCESS in 22m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/16eecead2d464ae0a11765f05fe6b7a7 : SUCCESS in 15m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/949ab8145fb84fbaab22ac03dd9ce5f0 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba5d540ab3bb4955875c9772965d17ec : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8d632ca5c25e462888909d839484bfb6 : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b2a2791dd2854fda8cf8be1da9f1a7ab : SUCCESS in 8m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f959c0ef51644a37bde919dcacdf1812 : SUCCESS in 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68ac1c8325a0422aa4184c22a0b40224 : SUCCESS in 1h 09m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bc6940232cdb43c785db7ad9f37472c9 : SUCCESS in 1h 56m 15s","accounts_in_message":[],"_revision_number":1},{"id":"23b1b5f68eef4e5ecab5de85c7d7d9b1f2a318fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a1aa9284850e02298b852019ca4f00bec6982f55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:04:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3b580e24464e76a69f8478ace921b0 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e9450ce58d7856df4985571c046078a616615cd","revisions":{"8e9450ce58d7856df4985571c046078a616615cd":{"kind":"REWORK","_number":1,"created":"2020-12-04 08:53:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/98/765498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/98/765498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/98/765498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/98/765498/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:53:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:53:13.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 2.3.0","message":"update constraint for python-heatclient to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If9554407b2a4b68e335749ab0e5e4b4e8e000054\nmeta: release:Code-Review+1: Rico Lin \u003crico.lin.guanyu@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I58ca6b0484ffd530e40904abf3372cc8483bfb95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e9450ce58d7856df4985571c046078a616615cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e9450ce58d7856df4985571c046078a616615cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
