)]}'
{"id":"openstack%2Frequirements~765499","triplet_id":"openstack%2Frequirements~master~I8a8ef1f7c10e40eb1252a3d1645a407c216167a6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8a8ef1f7c10e40eb1252a3d1645a407c216167a6","subject":"update constraint for keystoneauth1 to new release 4.3.0","status":"MERGED","created":"2020-12-04 08:54:41.000000000","updated":"2020-12-04 19:02:32.000000000","submitted":"2020-12-04 19:02:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765499-new-release","meta_rev_id":"bde47eb2a95683d6b13db846e2bdaf492873267a","_number":765499,"virtual_id_number":765499,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 19:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 15:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 15:35:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 09:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 15:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 15:35:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fd5f317b04c4488cb06ab4e559a43f135c842851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 08:54:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcda82b9d303795438bbf16b80d73a085510e0af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:21:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9df662c7dba74c5a9899de06854b3372 : SUCCESS in 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d90820183c04ebc1de36072922b6985aa34a3ad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e00b53fb788484a8380ad2ea6e6eb78 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0086c37b3d90486892d42cea42f1c9dd : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/558e83160cfb4422b97b338d4c1c4e1e : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d2522bf61984dbcb1df0335f803bb86 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5725b905be7e47209d45866671e67ce3 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e36e60bdf914932a19732e38d5a0751 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/774322e28dea4816aedad379124e8ba6 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9c538f44143e4824b863ce59fd9ba444 : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ada4d9237ff4e348f1d359712990515 : SUCCESS in 10m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/36fd0096ce404b848bf3f2a021eefb8a : SUCCESS in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b6f15659dcf84090a42a392524f80ff4 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/37564a46120d4bd49ff8b0df311a4b9b : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bf8368290984b29a4d25001737e4e96 : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a8d9f4929d384e34a42518ab1781c4b1 : SUCCESS in 12m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/be4a87665a7d40c9aba29fa6b20daee5 : SUCCESS in 13m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e920dcef81334b7d84fee90025c2fec0 : SUCCESS in 28m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20fcb36924284aa4bd5662d9369f2fce : SUCCESS in 13m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8204a99ecd74cfd9d32fbac32446178 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34ebd7fb9245437d9d9d5f9358b68d82 : SUCCESS in 20m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6948e4a6612745a8ad405dd3eed4e2b8 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df7b2b5bc5224805a72cc414c060a37c : SUCCESS in 6m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c15cad0b99f546f884e4a146677653e9 : SUCCESS in 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/238a02704f964c6687ab3c90121fe180 : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c675b962d2994621bcbd05f62361deca : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ac3a5c5cc10e404d6bbeef63f7bb7f871af48716","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 15:35:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5e7444fd79191b0ce280da4e949060ed240f9898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:36:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1cd13c548087e248ad3ba8cb17288a5d8ca8a8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f550cb0be24217841e4ea74a27dbef : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2120533a92c94389b784a5b925dca754 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1f5081d05ca4457a3b39a64a02d7069 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b74d3feda834dcd88a239f4e2eb35d7 : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b01b5d2582a342d4ba58c444f221d2a4 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ada2b538db4d4bf0a620722bbfae9a40 : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5dadf8c29489462d8c1c1fccc1f47abb : SUCCESS in 8m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4495efc13dcd4ffd8bdf650bd47e2b39 : SUCCESS in 6m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c46f705e3a6c4783963e39bdd35f590e : SUCCESS in 9m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/668ee8329405450cb1c3efca5c19dbd7 : SUCCESS in 4m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e841c0a76b3942a2b49ffeb45f070ae0 : SUCCESS in 9m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/46756d0894ae4486bbe10d86ee037ff0 : SUCCESS in 6m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2c59cdc3279c499eb72804382228d8cc : SUCCESS in 9m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f8f2cd3c2b534a1abed8faf8948b2eda : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/54efad6eaebd49888a0b4c112a61aa79 : SUCCESS in 23m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/bde95ac5f4384b93a39e1e8c3a8ae57f : SUCCESS in 10m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/191465b755eb4e7e87324006007d333d : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1daf565aff954a89953292147cb8bbe0 : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e3d19573708b4997b5c3c23a09559b95 : SUCCESS in 4m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4226b485085846f190ce6f15b45a10c8 : SUCCESS in 5m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/06dc470fbcaf48959812faff47f99276 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0c09bf91872450981b67a6fd13d7b66 : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61b749846de04650aced092d4944788c : SUCCESS in 1h 15m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bde47eb2a95683d6b13db846e2bdaf492873267a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56d5538a5a88766659478913d691a929e1131ea8","revisions":{"56d5538a5a88766659478913d691a929e1131ea8":{"kind":"REWORK","_number":1,"created":"2020-12-04 08:54:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/765499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/765499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/765499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/765499/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:54:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:54:38.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 4.3.0","message":"update constraint for keystoneauth1 to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I2b21e153a161f29fb711af875ddea7a2e40ca419\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Kristi Nikolla \u003cknikolla@bu.edu\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I8a8ef1f7c10e40eb1252a3d1645a407c216167a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56d5538a5a88766659478913d691a929e1131ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/56d5538a5a88766659478913d691a929e1131ea8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
