)]}'
{"id":"openstack%2Frequirements~765501","triplet_id":"openstack%2Frequirements~master~I6983cab5174ce2d7326a6113c0625e96cbe23718","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I6983cab5174ce2d7326a6113c0625e96cbe23718","subject":"update constraint for python-keystoneclient to new release 4.2.0","status":"MERGED","created":"2020-12-04 08:59:52.000000000","updated":"2020-12-04 19:06:03.000000000","submitted":"2020-12-04 19:02:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765501-new-release","meta_rev_id":"5da54c695588bf7d04ed7d7cc7f1f775d8d35aaa","_number":765501,"virtual_id_number":765501,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-04 19:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-04 15:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-04 15:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 09:24:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 15:07:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-04 15:35:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ec63e6d31433792a4c2c2428b763c5f809f9986a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 08:59:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50c689fc10395817fabe01f3b88b7eb5ef5eb042","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 09:24:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2c08939241004fd9806b639f57733a6a : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d9f5df3e18ed8dfb460fdae5dfee677c773f6578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:07:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d59d0d07f42f438b8fc89f94cc9d1098 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2db589bc6dfd4d8eb6e7629708d54461 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e8d05acbe940aa895eb1ede0bcba95 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ebd1b48876b40a49f168952aac01b97 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dff2d41593fe4fe78fba30d7031defcb : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fef3967221d4cd9810f88479ce38e83 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/84a7784a141848b39780cc016fd33876 : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/51a5ba5f76004d25aeb2690a3678417e : SUCCESS in 10m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/03abcd0228db4eb4b700d91daf43932d : SUCCESS in 7m 11s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a5863e0d530c4018aa54f75bc50858da : SUCCESS in 8m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/03eea872ae9c4301949453ce613dccc3 : SUCCESS in 4m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d120f215d0da4e2885e914aa5bcc0adb : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af28053cebfa463f92a2bc282869f4df : SUCCESS in 6m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ee956de0e0f841bfa25e647b587a3594 : SUCCESS in 7m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/78528825273f418f8adc6197b82d6027 : SUCCESS in 13m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6368a750013044a7ab0d2d9aae9cc2f6 : SUCCESS in 32m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3d7be567c5264af78ff91facad82bce8 : SUCCESS in 11m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2d45179cc5a4d8f8700859a39b117a3 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f42780d499547f3b34a16dd26a3937a : SUCCESS in 16m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/34c5ea705ae54b3aa814564dec162d2e : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98c5664fefbe4240982624beed5f71fd : SUCCESS in 7m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1a005b5b8f64451883f3e215b2b1f2e : SUCCESS in 10m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc2099071e8d40b0b12580c1e39b5fc1 : SUCCESS in 57m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d89813db67c44bfc81d5a1bdf70905fa : SUCCESS in 1h 24m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fb50606b4b60b78669f481f06cd5cb24a1f1e5a6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-04 15:35:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"086bcec7ba2237c5888dd19ab049018237d71e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:36:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb857675d55e2360a27f9f1d230b8c05256fba3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 15:37:33.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b297482bf4e2a651a14cff8bcdc7d752d5fb66f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f0a6b5b4b448a68d307fe9abdee477 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f817532f23054154b442c850e51ba606 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e5e772221341159466071fdb3f00cf : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2412857b8049433ca44a66ce807e1eb5 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f37382a9b1c498599c4fec10d6e25f2 : SUCCESS in 6m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/026c618bdc6c4edda7a4d5ef25bee25b : SUCCESS in 3m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f703c73da6a2452586e4122a664f02a6 : SUCCESS in 40m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1859838541944556bddd5e3fa8ae7322 : SUCCESS in 7m 09s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c72b0c4c513644589bb3a558cc08f39b : SUCCESS in 8m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ed87d87e7c3460285e0e02650e0fbc8 : SUCCESS in 13m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2f790eaf40e044e192f48ca83fecd94b : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb86c2276722458288e6f9a14532e619 : SUCCESS in 6m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6d1730a889ab42208b708890c0797f2f : SUCCESS in 21m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aeaedb6ab9744d7eac88a25cad0964a8 : SUCCESS in 12m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/794b09a653ae425a8dd30e8e8cb922b0 : SUCCESS in 26m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e0388628a5194a2f9ee8a270985d38b1 : SUCCESS in 27m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/500aa1e642384711a25853673c18e59a : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/235411ea88d64b58b01a0fbe3749bf82 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6ce682d15b7e44aaa773a4be9b1cbd64 : SUCCESS in 10m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/875e690278f743d0b0ac39d13c8478a4 : SUCCESS in 5m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ab35fc2bf0ba4943aefd26ba9762f360 : SUCCESS in 16m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/778f3e761c4643b6a7eca72486cdfa8a : SUCCESS in 55m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1848c70131a64282a0df73c7cc774fd6 : SUCCESS in 1h 49m 54s","accounts_in_message":[],"_revision_number":1},{"id":"da283bddd685256fb749e83b0145f46181f6f218","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:02:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5da54c695588bf7d04ed7d7cc7f1f775d8d35aaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 19:06:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dab9b5cbe79749a3a2df2b4aefca9177 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e6c6d5b4afb0561cf06118bddc8b4d4d361ce1d0","revisions":{"e6c6d5b4afb0561cf06118bddc8b4d4d361ce1d0":{"kind":"REWORK","_number":1,"created":"2020-12-04 08:59:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/01/765501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/01/765501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/765501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/765501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/01/765501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/01/765501/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:59:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 08:59:50.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 4.2.0","message":"update constraint for python-keystoneclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5cff8106f087f77269f32662b691598a8d3e4e32\nmeta: release:Code-Review+1: Kristi Nikolla \u003cknikolla@bu.edu\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I6983cab5174ce2d7326a6113c0625e96cbe23718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6c6d5b4afb0561cf06118bddc8b4d4d361ce1d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6c6d5b4afb0561cf06118bddc8b4d4d361ce1d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
