)]}'
{"id":"openstack%2Frequirements~765563","triplet_id":"openstack%2Frequirements~master~I50933e31d5b5201eb4e39cb3de6f07de18b135c0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I50933e31d5b5201eb4e39cb3de6f07de18b135c0","subject":"update constraint for tosca-parser to new release 2.2.0","status":"MERGED","created":"2020-12-04 15:59:01.000000000","updated":"2020-12-05 22:51:59.000000000","submitted":"2020-12-05 22:50:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765563-new-release","meta_rev_id":"ddf2dde7fa907e9adfe575ad85994a11891cfefa","_number":765563,"virtual_id_number":765563,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-05 22:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-05 21:17:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-05 21:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 16:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-05 04:03:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-05 05:28:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-05 21:17:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b1165cf7195a491c37ed05f75155948bcf438e3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 15:59:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33a65abe2c222b1c10a7e3e0325a0b8176496e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 16:23:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/743d5967227040108ac369258aaf266e : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"718fdfac165a5f7128c12518d4d9a55f980dff41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 04:03:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cff6b7cc5f7409a8f749a4122f9f805 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7972a32b5943ef83e1d2897ae85230 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07f6ec1dcd7a42b89350ddfa16702976 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/432a3f244562468588167eb2df16ee89 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55769432d1b749b4be8dfba2252ad984 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6db03e1e1bbc4e1db8e1236275b4b827 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5e80097bffa340949422d76df8b9a45e : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ad73f208b43a4cc892aff50289256b46 : SUCCESS in 9m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05fa9e6e511448edb549dafb9c4f2d08 : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7dfaa47e767142fd8a4cfe65f999be69 : SUCCESS in 10m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ae448339eb1440f1afe9e279747d2140 : SUCCESS in 4m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e735f2389b84416ba8020d69cfed6c1c : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/656f29d8d158456aa60f9df982540874 : SUCCESS in 6m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dbfaa3050f09409b956dbe715c00a9b7 : SUCCESS in 5m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19d8b88ec1bf42fbaadf2e9f8c77fc31 : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e1f6c587885d44ee920fb16776102b0e : SUCCESS in 31m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e08c53f3ca0445ab51d405342fe5142 : SUCCESS in 12m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecd08a2a75584d8c9d2407e97b216994 : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33a6573097a44fe9b7b44d1d7cd3bb4d : SUCCESS in 19m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a5e1fcd1c64948be93f9e883d649a766 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec06a46a9bc042f49c47561c9f1ead49 : SUCCESS in 5m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/411831b1fd4c4affb9371d9ba33c267e : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57ed924126c0462da2d8d1dc5e8637d8 : POST_FAILURE in 28m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b37b056d994492ca56e12180e2a1f3b : SUCCESS in 1h 29m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2c93af85804f8fc6c900361beacf86f58f375aa1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-05 05:28:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd7742fd4e4ade753a503844a70b978c2351a0d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 05:52:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b82d7d98e3444f3ea3863c59b02afbd8 : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0c9d9ab76cee304ec7fe72693be1c68fdfb8500d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 08:52:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f0c4ad6946d4b7584e027ebca9dec6c : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9996236748194663bb9cc17556263cab : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3b56b7dbf984b518f16f2869f3ae52c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122062bf943c4729b8415ba23b183af0 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12f2d42d290d427890bc5b517815f627 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce1e19c5ce0b433895a19051ac62bf12 : SUCCESS in 3m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4bdae94e7cb4503bad24ac208615d42 : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9e87c9fe46b4e66bc249b253ddb64d7 : SUCCESS in 7m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/46564907cac043be8632afd2dfcb751b : SUCCESS in 7m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/adb5fd611ee44d6dbeb07973eaa12a3f : SUCCESS in 8m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e56f1bab84a6435189bcd65047ad0925 : SUCCESS in 4m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/86d892f930ba4d36944d46fbcb63a35b : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f6f2ffaa928d412e966f3ddcab12ffdf : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/312fdc0fa56c44978422f44b8abcf8d5 : SUCCESS in 7m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7e533291366e44508a9d0ec20908cf0b : SUCCESS in 12m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/31358ba273ac479e8155e1350f2d2929 : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fc3502f228454648bdbc90139cd362c6 : SUCCESS in 12m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a9b83f4fc7224fa88fd5eeac0b712433 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/202d394f410f48ba8a578b433960454e : SUCCESS in 21m 31s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57464709cb4649f68c9608c399f89f82 : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c54db6a406764f56963b93e009fa7e44 : SUCCESS in 8m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5cd0b555d2c9430f9a1ccf311340d97a : SUCCESS in 15m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c67c4e90d2a4707b895c020e148e8c5 : SUCCESS in 51m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/836c71978ed3489b8c9f8cde97af9ec5 : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3229321493c16662536502728bc2858c718a17b8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-05 21:17:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32d94e11149e38a6882a639db7e20e6a2bd27c4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 21:17:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f056b1719926a0e07f02d730b21096f8e87a178c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 22:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c227021bb374aa08e634961f93ed1f0 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c8e03bf4814cd8899fcb9c621bd0bf : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bab5fabcf16f497b9fbc1f7f586e8a0d : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc4e4af0d9c64521bfa401857c1d809d : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e72e57541aa4438891b79cd04d0b2692 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b56e18f3e75400ea711b53097ed0c31 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/86a01797735748088440188378563d3c : SUCCESS in 8m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05de1031057f4babb6dc417d9ce173d6 : SUCCESS in 9m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd691cdc394d47eda186944cd691ab47 : SUCCESS in 8m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d6dcf3b239e499fb70e1e30461d64cf : SUCCESS in 5m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b9dbe50a621e40b2b88ae2c882a3fe4c : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/50de6f6e83664ad7a1d6cc28243d9d70 : SUCCESS in 6m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6293adad1b014bd6b8901f39ea2a74ce : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1bd0ff702b75422683f61e07b33c05dc : SUCCESS in 12m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/86fa407bfebd41d89e3a1394c708cd82 : SUCCESS in 25m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1ea8da4ab5c44b418346db0f28583565 : SUCCESS in 11m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b35c2c6fd6b442da89d38ab56ac826ed : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3efa36bdea7f435c856bf5c4dfdd0552 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/edd7fb8184524d49928e81de4168d1ad : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17ab8e91030d4ad8a035a5b112e7618e : SUCCESS in 6m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d96e82fed67143c58c0bc50f28fd6833 : SUCCESS in 10m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a0353ee40a546eb99c8121e2ee681f9 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/631cbba4fb1849178a0f043cf4457210 : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8e2774c96cb255e48f85d7ea938911cad236e0f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 22:50:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ddf2dde7fa907e9adfe575ad85994a11891cfefa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 22:51:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee5cda7b6b64edb899009ee2993002e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9cbcfe5c792caad61fb6bcc06f280f119f4210e","revisions":{"d9cbcfe5c792caad61fb6bcc06f280f119f4210e":{"kind":"REWORK","_number":1,"created":"2020-12-04 15:59:01.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/765563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/765563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/765563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/765563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/765563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/765563/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 15:58:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 15:58:53.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.2.0","message":"update constraint for tosca-parser to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: If82ce7774dc3e494e1926bf1d28f11d6fdb1763b\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I50933e31d5b5201eb4e39cb3de6f07de18b135c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9cbcfe5c792caad61fb6bcc06f280f119f4210e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d9cbcfe5c792caad61fb6bcc06f280f119f4210e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
