)]}'
{"id":"openstack%2Frequirements~765576","triplet_id":"openstack%2Frequirements~master~Ib51c9fcc5af74305266473c1c5ca06a4e5642f5e","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib51c9fcc5af74305266473c1c5ca06a4e5642f5e","subject":"update constraint for python-senlinclient to new release 2.2.0","status":"MERGED","created":"2020-12-04 17:57:51.000000000","updated":"2020-12-05 07:11:38.000000000","submitted":"2020-12-05 07:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"765576-new-release","meta_rev_id":"8d08c6c8becee94661bfd33534a285ee853120ea","_number":765576,"virtual_id_number":765576,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-05 07:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-05 05:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-04 19:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-05 05:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-04 18:30:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-04 19:07:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-05 04:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-05 05:28:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4cfbbcecc94290008fdcebf316e03d3e31d8223f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-04 17:57:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dd5ed4b5f4ab01a97548783128b4e243582f3ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 18:30:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f06716f41973449faf9ecaed88bbd0a6 : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":1},{"id":"81612fc09c16eda0a7e7f226b74e3f626590f722","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-04 19:07:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecc2a12115c63bfd1ece47050832995d5c8a0e47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 04:55:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ebe382207d945a3a128d77d332385fc : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29566d1d54e49a5a1057db7cb58f33c : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/615071afd6254374aec4485a48744196 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b54174cb32f42caa118d0bbfe35a7e5 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d1714332dff4f22a164cddb32aefd07 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6712944d415d4d74ae5222d9a2113062 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/925dd447837541cc9cbd262becbc2e19 : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c9e096013e2463595125561e9422aee : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a262ae8d7a5407f8cee245b2aedfb01 : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/62915f97b313445aaacfb83e476e38e6 : SUCCESS in 10m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/11f80d8348604bbba253b3f2d4be764a : SUCCESS in 5m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3c492c9e65274fbe96e285be9da8cdc4 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e502b2ec1dc4ea7a7fffd994b1e5351 : SUCCESS in 6m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/db884953024b48c596f4fa71da06da4b : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2cb3db1485ef4711baf67883d7abdcb2 : SUCCESS in 12m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/830375d90c694de39ec5c524bad7ca88 : SUCCESS in 25m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9a78a31feb3e411e9623aa0ad6f31174 : SUCCESS in 12m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c451e171e5a4e4b9cf5f5fd14bd4d47 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9043ce861cec4793957604df2fb44a19 : SUCCESS in 20m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c8eea6b5d82043ac9154f8bee37a8000 : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6696f5f7274c4e2b8f4f1754880af1e9 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9bc1deae5c0a414da224588b377b7488 : SUCCESS in 10m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d569ad330778496fa271041f805d3176 : SUCCESS in 52m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d91ea01d0d144ba9135f2f6bb6e1bbb : SUCCESS in 1h 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"94b1c254d9732c57768781b07f4225b0aaf3a8cb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-05 05:28:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b5a2594761fc146d623347cd747e5df29b1d6fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 05:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e132889322a50b471f7bd3d9bebf51d3a055155","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 07:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6cd41ac77d4f8187411c865fa2787b : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f098bb0fdb4d453fbcbee2b89eb1d9dc : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f04820959f4ed99f5f4e5f01a07747 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dae598ab795c4c4b9a01aa070e9196d3 : SUCCESS in 3m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3a16df782a8474f8bcb7ee79070404a : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4af962584a044c07bb663ce9b97f83b0 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/64fca7ddce8b4783b267f218915369fc : SUCCESS in 7m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/05a2ecb6712140d3ac5d1fd5a5ad6e2d : SUCCESS in 7m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dcbb2bc5a3bd4bc7b9eb0c00483a8748 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/227f0d896bec4297ae3454361ea2816c : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/68d435f1280542adb9088bec00bf02ad : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b680f96758f494b98422d3c05a13943 : SUCCESS in 5m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3a8a3d41a0a44ca7938e101c556cd6aa : SUCCESS in 8m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6f035401afc84edfa663f836a7360573 : SUCCESS in 13m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7304fa3844e946fabb5c05c0da4ee64f : SUCCESS in 22m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/865a23f21e2449b1855b99117ac40a1e : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c56348cf50c042d0ace18eccf1b1c37e : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/331daaee9d5f41d295c933edc441c028 : SUCCESS in 16m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2beef17a207346cbaa1318a101b21b6e : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9063a81655a94b6286523a4ec9a33364 : SUCCESS in 7m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/63e69d09a00144b9af5d22549965eb5d : SUCCESS in 9m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee3b1ed204324fdc998fe56e2c46f178 : SUCCESS in 59m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dcc036b359243c9a6962a2b04473292 : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6cf69441200f0333945a130bf59f320c13dbb96f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 07:09:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d08c6c8becee94661bfd33534a285ee853120ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-05 07:11:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b45619e2db94e3e9943e0b8eb024eb8 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"545109a74a78074d11e2168e95c765838ce83041","revisions":{"545109a74a78074d11e2168e95c765838ce83041":{"kind":"REWORK","_number":1,"created":"2020-12-04 17:57:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/765576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/765576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/765576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/765576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/765576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/765576/1"}}},"commit":{"parents":[{"commit":"ef6f6e850b731719c6743b4d43e4194b32319beb","subject":"Merge \"update constraint for python-manilaclient to new release 2.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ef6f6e850b731719c6743b4d43e4194b32319beb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 17:57:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-04 17:57:50.000000000","tz":0},"subject":"update constraint for python-senlinclient to new release 2.2.0","message":"update constraint for python-senlinclient to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I75f5db718de9b7d542c5fc74724b3c3c717847d2\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ib51c9fcc5af74305266473c1c5ca06a4e5642f5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/545109a74a78074d11e2168e95c765838ce83041"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/545109a74a78074d11e2168e95c765838ce83041"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
