)]}'
{"id":"openstack%2Frequirements~766032","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I8bd114bd3893663232db74157e1aecac8bbb8b6f","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I8bd114bd3893663232db74157e1aecac8bbb8b6f","subject":"update constraint for os-brick to new release 3.0.4","status":"MERGED","created":"2020-12-08 16:00:04.000000000","updated":"2021-01-08 08:31:59.000000000","submitted":"2021-01-08 08:31:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766032-new-release","meta_rev_id":"e4a435d350b88e38ff86f5bffe7e023e5ef98340","_number":766032,"virtual_id_number":766032,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 08:31:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-12-08 16:04:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-08 04:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-01-04 15:20:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-08 04:13:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-08 16:04:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-08 17:45:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 19:10:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-22 09:03:11.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"CC"},{"updated":"2021-01-04 15:20:19.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2021-01-08 04:13:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"918373547ad3987a129ff05563a1305dcaa6842d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-08 16:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1fc1f33c88f64dbc4bd85bdd75f10f5ce4a3afc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-08 16:04:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9995c9b963cf4904473ede23fa5ecf8ec59094f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 17:45:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027632f593194f5ab83ac80ddaa853d4 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc63c717441248d4865df44d07fe7fea : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6b2e8a529754c799f048d415047af12 : SUCCESS in 3m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b043398757f14caeb91abbacca76aa4d : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/491d793b67d74eae9a8f7dc596e30ebb : SUCCESS in 3m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9b38b2c3570f4db18d511db10f099bdf : SUCCESS in 7m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/df2744c9c2ba49598c736fe6a3500e8d : SUCCESS in 9m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/84b832f7f3cf4d9bb61f5d7b0524ce18 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c540c4938c1d4e0c9afd6082d96daaa6 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e7256e8c31da4effbdcd89a4b48e83b7 : SUCCESS in 7m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fce855db2f9647de98dd1afc6682c40a : SUCCESS in 13m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/356754e827494fcf8d3dfaff5b03b9c1 : SUCCESS in 28m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eeb0bc68262f422993744308bb7891d1 : SUCCESS in 11m 58s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e5768e3bde0e41388303bf3217f4e246 : SUCCESS in 15m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd8f5ec713b64b1c8f5876aeac73b099 : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1921b9947c2e4369beff8fb8405a9637 : SUCCESS in 5m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b45dde010c5f40d0aad07f9e1d5bb10d : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/587e9d0ecc8840a984912e1c20a953ca : SUCCESS in 6m 50s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/edee4ceafc9549e9b076f366a1669292 : FAILURE in 3m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1596570558247488035f92d0bace4c3 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a010380e9dfc434e9655d30816aacfb8 : SUCCESS in 50m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3f71053c53e4f51a1489e24b28c7bae : SUCCESS in 1h 11m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/780d2d0316fb4f38ad2ed9c86dc0b5bc : SUCCESS in 1h 20m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0e9d1943b54d4bb69edd9cae1d0d67ff : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"42cf1a3cde54abd944bf88cd246611e1f70b26fc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-08 19:10:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84103a514be9d41c9f2e057bb4caf6b9a5863c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 20:45:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a8eeb54864478198fc7f63eb161ba4 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b30d80eb409d4145855e6457dfc7b9e6 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c555449dc124b4ab03d6f6a21bce02a : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd580557df74496882a5faddbcb2c3e1 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fca7354facd4f29a2302728bca2e7f4 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42507e4080be4c03a9dd38773bbed478 : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ea5651f8d6fa40ea82b0da7ae60ca1f5 : SUCCESS in 9m 04s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c438b1181fb4fd6b38c41a22892c73f : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e55f47013c264f089b5e514c72ad639a : SUCCESS in 11m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/065d6e4a00c2444da5ec7a95f261d6be : SUCCESS in 7m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f84f7a166ca040e5bbfc508361a71767 : SUCCESS in 12m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/22519e0d5f444a559af770c7c45508ee : SUCCESS in 30m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00ae1da539b544a099005b06d5913f6f : SUCCESS in 13m 16s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e4af9c2850244eeab53cd5e7a68b1eb6 : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b15b6928cec4482969e5100552bd649 : SUCCESS in 20m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/220a95c2a3f949ee8167a3d9948581a7 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/be58cd6d5f9e41b888e55a7744ef8c13 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0130c644e7f04e9ea5dfec6a0a158094 : SUCCESS in 6m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/16bed9a39ef441de922b32980667abc3 : FAILURE in 3m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2610c65be352435895fd12638224f70f : SUCCESS in 9m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/deb05da2ac994b67acbc295190fe018a : SUCCESS in 57m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cfc0167425254ef4884898240b0ca2aa : SUCCESS in 1h 18m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71fbba7e8aa246e08e6a64358651906f : SUCCESS in 1h 29m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2c46f40da597405da95619f653927f7d : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":1},{"id":"7ca33ed417c9a8c05bcba8ca324dccccf510aa16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-09 04:21:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bfa94565141fd35c33a9a9ff1b96c51a32aaa0b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 05:54:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c3e00df35849f1b6f14e777b605746 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb803089f36144458e3736066df6e3b6 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3da3f216aa3f4c21ab57d0ca53c5d6b7 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84536a430faf4a3fa1a70e7021747963 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6bfb6bae873a49999bfdd7c9fb858452 : SUCCESS in 3m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ceea593f724c4798a4a387a88d9377be : SUCCESS in 7m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3ed7323085b4d73819f17570609aafc : SUCCESS in 9m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/943aa0e2b54d48489d11cfe7b7ffc652 : SUCCESS in 4m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30f1573b962e4da58fbfc68d5611ae97 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ad1c66009c44299bfb6fda682c6ac06 : SUCCESS in 6m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1b260f2788e4fa790116d2777e0e3c2 : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/73225c5234f4477aa6adce45ff128385 : SUCCESS in 29m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e06757487d3d43d19efc284ff800187d : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/471b00463c4d45ca8cfbb8715cfa59ec : SUCCESS in 13m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/88bfcc96cd9349a0a296747d808dd69d : SUCCESS in 18m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/964db0fa837e46a9841d69dbabbe66cb : SUCCESS in 4m 48s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/0b5aa7dfd704410e9fcaaae7c541de26 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c7b713100415469bab64b4dc20019744 : SUCCESS in 5m 38s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d36a8e4b4a374b1cb9a3a71c85c37e48 : FAILURE in 3m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/18f09aac7e0d4f6e94744b9e8438a736 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05052bc69d3a4507a2a0252bdd661d4e : SUCCESS in 57m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1938eb6390c54df2b369c4c3573cedef : SUCCESS in 1h 26m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/886ee791e8174b10a8b1eb28d64a50f9 : SUCCESS in 1h 20m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/5afbfa49059846a492fe365d8fcd9dd9 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6e2e25cdf96294bc6dee8b721fa707983bb3a658","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 06:30:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2f0fdb04262c12d06bd856018be6ddaf9012a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 08:10:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a281d0d0984671aef10188989f6040 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cabd579971c04df5a38840314ea620d0 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33132e1034124fed9a7d2d83d5d01f29 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bfd28defd2504998820acb4a42e527d4 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cea7d8e627c046a19a7e78239c1319a6 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60a90ed0963844a4a07fc3acece789fd : SUCCESS in 15m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/50436f8f70814f2a89062770f06f2c85 : SUCCESS in 9m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fb7516bbe23e4e089123d2702c6ab073 : SUCCESS in 5m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8d278b1a353a408eb66b5d4347117409 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80545e4bb2d84a5da54bf2212f46f55f : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b55fd583ac046debc1cb4e79e817c95 : SUCCESS in 14m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/21875336261a472abd6c6ac4ae3a1ce8 : SUCCESS in 26m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd9a1ae316a74f449c15333909b36051 : SUCCESS in 14m 37s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/b67421f244b74d29868bbb5d73b4e74c : SUCCESS in 14m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/108b9dfc102c438693906f36d680673a : SUCCESS in 19m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0cfa040590ed4ceeab3bb29ac815b855 : SUCCESS in 6m 04s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/2e882731e6bd4d36bcff422d056010a2 : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/757fb59821354e2196ee2c5a7b357960 : SUCCESS in 8m 27s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/dd1a6afbe12341688739ba1cef2c2654 : FAILURE in 4m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/306ba75466f84e1c820577f5474da4f6 : SUCCESS in 8m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f589494514543758cfac5d41d90f16c : SUCCESS in 49m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c322e0a558084e91891893f652a7221e : SUCCESS in 1h 17m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d4645d4544e457da77660173ca8d3d1 : SUCCESS in 1h 33m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/1ec1831925424a00b457f5ac4e60513c : SUCCESS in 1h 25m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6819e52d40d0e85d20fb5bdb31b28c2306d251fb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2020-12-22 09:03:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7859c06e8326e473f106ef0dc90ef8d8b4ef51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 10:32:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12bbc1409de34d919c2ef4eb65835205 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7515933fb9b344749b6e8a6cd9e4586f : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f8b9dd912824661979216a74a9bfde4 : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/83fd5d7a79264fc39ed24012b3263a99 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b18f1adac3ef442f89873ed3e42c760e : SUCCESS in 3m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/42596bc142994f6eb60cf4d20ae7261d : SUCCESS in 10m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ebf77cf73c1243469cb75af8f2168505 : SUCCESS in 7m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f9db4831db8453d8ab53c76ad7b0789 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/efe14516af22447e8dccfce939ba7894 : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1974da3a66d7488f8e5e1bee92675ab7 : SUCCESS in 7m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/604d8cb0e7d74bf09658007ebe965048 : SUCCESS in 12m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/80639abeb4714b69b8cf6d8aa519cefd : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6973f3944bfc4f4b9ee4b35baf639cb1 : SUCCESS in 13m 10s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a91d187f0893493aaa285339854a1b34 : SUCCESS in 13m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14b2958ddaa34996bbd1925537cadc68 : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/61fadc09fad24305ad8dd80b1cd02869 : SUCCESS in 4m 47s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/6442ab301f444d6caa0263e6cf33051c : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cbbbb303c7da4307878bed162966ccd1 : SUCCESS in 5m 54s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6ad78285a3ae4985b6fcaf185f80255e : FAILURE in 3m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/813ba4938ae24909ae06f3f55879b428 : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd9785d05e374770ab682a036e18622d : SUCCESS in 1h 01m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/51cc32c53faa420ebc83d3b6ce1a918a : SUCCESS in 1h 21m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d85cb87af480429cb0446a20212cf116 : SUCCESS in 1h 21m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f29390fb909c43b2a7d2cefcedca49e6 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b6e3ba9519e543a0e549becb779c226606c32306","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2021-01-04 15:20:19.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"69ad1aa97fb76d77becc9fb8c9853462bffb43ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 17:08:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1388b9cbff344f43ba08add48f289cfc : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/571311d2485a4a05802006108788b8c0 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a136fd0b3b749168b671588c46e9f4f : SUCCESS in 7m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac388a5b06ad4cffa8275e455b53c411 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4223595f6254a7c8ed8aa943dec8104 : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/32f78aa6e533476384106b65633c6b73 : SUCCESS in 12m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/22620d326d0d4e3b849c5613059a537a : SUCCESS in 9m 42s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa500f18c2e640659e8fc58e2711f749 : SUCCESS in 4m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c05babf9d98b4e30a89aaa17ec3d963c : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee0260bee56f42cd90b4c7790652cc5f : SUCCESS in 6m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4df68b8e802e4a6dab50e67d97ded333 : TIMED_OUT in 1h 01m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d171238c8ba2439085899c16e9ff3451 : SUCCESS in 28m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0d55445bd80a40b3a4b48fd240da38d3 : SUCCESS in 12m 48s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/858e2c033d794d7babf5462e804141fc : SUCCESS in 20m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f14ad1277d44877aa91b8c72d0cd1f3 : SUCCESS in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/98eeac81d3a149edb4a98a5e9eba30a6 : SUCCESS in 4m 45s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/b9debe67239c4b7d830b1e125e3ea4f5 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89fd7e9090bd49e685b1aa3802684967 : SUCCESS in 7m 14s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/6eace505bf1f472f888aa64974b68779 : FAILURE in 4m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b4aecaffa88b4d4599bb1d3896d95b93 : SUCCESS in 9m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed39e1ceebed4e549be41bc32c0772b6 : SUCCESS in 49m 16s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/347f0cd13b73434b8a50c5b6a0fb00c8 : SUCCESS in 1h 15m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3a207a7b3194c38bb142504c4e3d0a0 : SUCCESS in 1h 23m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/3bbdb6555403454486399447763d0401 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3700f8eed50500674e8b57e98f4a94a2c4c7d84b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-08 02:28:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"42eada9491eddecdedda52cda68802e95e2798d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 03:57:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb5daf9eb6d4cccb5faa93e57d1ea5a : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced0d2d9e99147f2921f1ff4dfe85b4c : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5f752781a3d4ffd8c6311c01d144b53 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45ac6e9a47f94fdb9a9e6f40dec58c5f : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b1beee169a849a0b47e98f46349e207 : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/de6135f151f14d3c80c4b42dc6f47be0 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d962a1c1c7794eb8a8cce263350aadcc : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cbce0928fad64a0497139f727451be16 : SUCCESS in 4m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/829b93a41e374f969872bd4d2855cc6c : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/25c5fe4740a648f1a74b06156e252025 : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2b833da7ef41424ab04b6cc773c79f49 : SUCCESS in 12m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e87a049436f34e16892811596e57fd4f : SUCCESS in 27m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/608ba5989cf8476f9fe773c4985120ea : SUCCESS in 12m 56s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/5ca25564b1e24a90aea152191678203b : SUCCESS in 12m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e76c6b93c31e4e258759df443f874554 : SUCCESS in 12m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1cb0aa534c5e489794b2e8e3f9fe6a6f : SUCCESS in 5m 32s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/f7079fe1257641d7bb64d2893e0a0c04 : SUCCESS in 4m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f58e5f3ee2541959cacbad2b7b3eb38 : SUCCESS in 5m 33s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2712c1aa532e4498b4a62d396820ddb2 : SUCCESS in 9m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/edd8af8cea0d4749abd1283bc0ee298f : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f63fc9e1d654f6f83a2a5aad46f3925 : SUCCESS in 39m 03s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/127312717b6443d39d932dfc72f10ba0 : SUCCESS in 56m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/984ce938378845e19ccebbc90f8a2b5e : SUCCESS in 1h 22m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/bafcc44594a2437cb1a186d28f8f2366 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a820c00c8f9b50cfe11cf90d838d446d7be61b9b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-08 04:13:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49ba6a4074266c032a5ae845f2394277de0991dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 04:13:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"165c3c53a85de335ce9b70e3b38ae00a7d7ce95e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 05:34:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66cbe54867d5498780bd4998becd41f8 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62f5c1d6bd584ed6ad7d0396e919879f : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/44648a452c684e5097b42340b1b50fe7 : SUCCESS in 5m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f028bcefc19b43bb9a4f14ca3e4b37db : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8d11ba366184554a2c4ffc6fe7ff634 : SUCCESS in 10m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9e71f2ea755c46a1b2f40ee7c0adee31 : SUCCESS in 20m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/66bdf3697c8546e29dc3c0a2546342f3 : FAILURE in 23m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a4e6177dce19497daf93a2539599d1a0 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5921f36a79ea49b3ae7e98a16373cd7f : SUCCESS in 8m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/51f48d67fe254870bf203a5d91a67dea : SUCCESS in 22m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e3a626fdbac84d2087bae57ba4d7c673 : SUCCESS in 15m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/65032995f346490fb3941a21ecbf6c87 : SUCCESS in 17m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ca8ada0b7cc47b3ad169e587558ca06 : FAILURE in 36m 52s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1f03154da90d4262a782b90f0b71696c : SUCCESS in 10m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c56e9bd82d15498ba644c7f6fad1cd67 : SUCCESS in 14m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cda8e2c6680e42439c15581c1ed57479 : SUCCESS in 4m 26s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/81327bb2134c41cfa7b572ae54b291bc : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/735a0dd50b0a4bfe85dfac999e96f510 : SUCCESS in 7m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/93eb02f47049496182a81f921bdaaf93 : SUCCESS in 15m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a28ac4574d5a4292ac5d533390cfef04 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/628ce0a0c7694b0382d975b6b28ec880 : SUCCESS in 51m 10s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e65465d3020f42dc872cc9d7c29b07df : SUCCESS in 1h 11m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/89170b8465fa414b94a5179fa1bcd1c1 : SUCCESS in 1h 17m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/1a4dd67449af4b40bc5ac2ff068043eb : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bcd6b90ace20a07cd53f66196b9aede5723a5a87","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-08 05:42:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d57c0d5031590497bc1cf60af4c1e105e8c03248","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:08:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c030486e5d42446ba2bc81340878c36a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f09e07d79eec4cdeb7680766277db710 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53bd7931223c4cbfb2cb66d49e0fef16 : SUCCESS in 3m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a380fa9df844006b42caf2b7f7e4ffe : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46e60f377b9d4ac0a8b0ff1173c62969 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/464cac07071b4ddf8f67099b3ca1c626 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b5458a73555a47b3b7edd71e31353cf1 : SUCCESS in 9m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3072f6749d604d8cb5d0a42874e7f911 : SUCCESS in 4m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/596f9b4e68e94bd0a6f40733aa6a5c63 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38a2514a0a3b47a2aed356c429da2491 : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15d40fc594374cb5b4396be5e0545df1 : SUCCESS in 13m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7917a140dbca4cd69fde3b7029723f1f : SUCCESS in 31m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/79028a180f5f4440b024d893e19e1a5f : SUCCESS in 13m 05s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a60f3e0dd7a64b9d8a8ebae59ad0072a : SUCCESS in 14m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3aaf959eb3914ce2841f46ef600ef778 : SUCCESS in 18m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d1d3a474681d4813acaba2967fc86bc7 : SUCCESS in 5m 08s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/7ea8b86d4f154a6aa1dff4ab1b74ba22 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f64cebe9e204954ac872264a41d891d : SUCCESS in 7m 43s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5b5c42a7c8f14ecc831fe0a06eb5edeb : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ec870324aff049cfbcaaf9545092df0e : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11e5e4ae3324496db764898fed685d50 : SUCCESS in 56m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3082e120e2064272829fec4edf8ed391 : SUCCESS in 1h 20m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f808c34a7ff407596c5d2e037d517e3 : SUCCESS in 1h 24m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/4a6d01252bab4b34b8c09779d45f2b5e : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0348e57bd066d30e00d7d92c565199436366b577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 07:08:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f857ba820df119f3917a4bd4071d7890d97bd74d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 08:31:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa85295a9f1d4b3fac382fb1e95feaaa : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c391d820e004e688c50e9e4711a7bf2 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c1c480d63894b51b04b7fda8adc279f : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0922944e9e674007b9afe560d27f0425 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/208822313c524048a112ed5c86cff150 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fbacd3a8a5524d0e87740d074b98df0d : SUCCESS in 15m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/596f5a4d81ed4d6fad29ea081c03f82b : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5db054e4d6ef43348345169ac00bd6ec : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/03bd3b77cdc54cbea277bfec3faabd83 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/494ca307f7fd4ed4aeb186ca551194ea : SUCCESS in 7m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e4ed17cc961d49efba4b2dbc7911c314 : SUCCESS in 13m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e86844a893ae4ad1b49762966104659a : SUCCESS in 22m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dba3f7ab3b5f4def9e505402569be624 : SUCCESS in 11m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/79d4a816d3144f2dbf79f6be3f3cd073 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/61e39af5af144cc29e8e8392682a5d50 : SUCCESS in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dcca84cba5c340a58be13f882fe1aabc : SUCCESS in 6m 36s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/014480ba0b654ebcbbdbc1e009390076 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/88e1ab036df749d1a9a8c37d6db116e7 : SUCCESS in 7m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/76814b29bd434905869e5371b65b5ce0 : SUCCESS in 10m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c870ff2183a247e0b90cbbb865c2824c : SUCCESS in 9m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0e785dccab4a48831c468e37ab294f : SUCCESS in 51m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/66edaa9543d2402dbb0be2d3ca3f0456 : SUCCESS in 1h 18m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1af0f313b4b34515a6014b17c7730368 : SUCCESS in 1h 04m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7e060faf3a7741a09f14f8201336abc8 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e4a435d350b88e38ff86f5bffe7e023e5ef98340","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 08:31:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8016ad740ca6d72cfa1afb4061e82227285a54fc","revisions":{"8016ad740ca6d72cfa1afb4061e82227285a54fc":{"kind":"REWORK","_number":1,"created":"2020-12-08 16:00:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/766032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/766032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/766032/1"}}},"commit":{"parents":[{"commit":"b33b81b3bbe4263fb26748fd05a466e4a83e203b","subject":"Merge \"update constraint for castellan to new release 3.0.3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b33b81b3bbe4263fb26748fd05a466e4a83e203b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-08 16:00:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-08 16:00:00.000000000","tz":0},"subject":"update constraint for os-brick to new release 3.0.4","message":"update constraint for os-brick to new release 3.0.4\n\nmeta: version: 3.0.4\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: I922ad4fb5c5fe5eaeee6467c4d55fddef4fabe86\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I8bd114bd3893663232db74157e1aecac8bbb8b6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8016ad740ca6d72cfa1afb4061e82227285a54fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8016ad740ca6d72cfa1afb4061e82227285a54fc"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
