)]}'
{"id":"openstack%2Frequirements~766170","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ibdc34ab30314d4ed6c3133715d77486b92ac71b9","project":"openstack/requirements","branch":"stable/train","topic":"bug/1907438","hashtags":[],"change_id":"Ibdc34ab30314d4ed6c3133715d77486b92ac71b9","subject":"Cap bandit at 1.6.2 when using py2","status":"ABANDONED","created":"2020-12-09 09:55:24.000000000","updated":"2020-12-09 15:41:50.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f205917d83e37b231563d13cbf5e827134bf43d2","_number":766170,"virtual_id_number":766170,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-09 15:12:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-09 15:12:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9430b30f09bc220d7593720f47228ea7b0a19f09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 09:55:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f6ac5e518c51c26e12b6c06a4317ffc1e904c86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 10:08:48.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"ea25a474a3b4b7c958f70f628761b9a3f297c001","tag":"autogenerated:gerrit:restore","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 13:43:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"02958de631a17c72f695ca2e266f727bd6d6ed60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 13:44:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"64d90525a2f2d78588c0893d3324e3974c5a812c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 13:55:35.000000000","message":"Patch Set 2:\n\nhttps://github.com/PyCQA/bandit/issues/663 suggests this release might actually be yanked","accounts_in_message":[],"_revision_number":2},{"id":"12a952d486886e28f6323fa108380daa3cb16d6f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 14:16:30.000000000","message":"Topic set to bug/1907438","accounts_in_message":[],"_revision_number":2},{"id":"fd41ce663a260bf0555d3c45982c8f899329e8a3","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 14:16:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1216182e2044269d206122643ad7fcf6afbd72fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-09 15:12:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5425087d76b4530a519e27904ad18f7 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7f3817aeea54e2ebc55709288b05d19 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb5ed39665ae415097747190eef45290 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db161e67252e463aa037d68c098be8e5 : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6519eb30d6d040c3b2bb1df5c9dd00f9 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b34ac0808bd94eed8b08e6bd213569ee : SUCCESS in 5m 19s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/c1e76cdb6add4129b1a891e2b2aacf8e : SUCCESS in 11m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/56f03916e2504511bb70696c1d78dbb3 : SUCCESS in 20m 24s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/042fc04bb95942738885fddb86753ff1 : SUCCESS in 10m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4cddb5d29c424c4796e2743ef79bb170 : FAILURE in 5m 12s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/54df3f70a0db4188bca511a353ce75dc : SUCCESS in 6m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2dd3589e720445b4a868a2d970599dbf : SUCCESS in 5m 53s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/f38c746842f747fcb4a25384edb478c5 : SUCCESS in 11m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/46805cc19ce54d578065bc0aaa3b295b : SUCCESS in 8m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/59cee5f09e74415282087c16206b0284 : SUCCESS in 11m 23s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/39f197a48ee94f4781424786f8995f1e : SUCCESS in 16m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0490c150466e48bea4f303f957f55865 : SUCCESS in 14m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/dc045d6639ba4f1d952228a649023433 : SUCCESS in 26m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a52586808fe64c34a0b75f65570664c3 : SUCCESS in 26m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3a825059c4b74a50945355f46106248e : SUCCESS in 16m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a7f9cfe3081646f5aef27a1f69f9a688 : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/e2e856b036b0443182143e8a92883701 : SUCCESS in 14m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d9ad7182a5f84839964b1a05e4b1bcef : SUCCESS in 18m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d43b666f8bd44fbd88c08244d58c6eb5 : SUCCESS in 4m 59s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/cb083a29a58147afab879ea7c7e92c71 : SUCCESS in 8m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6086c9e19101452ca1ee15a756085450 : SUCCESS in 8m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0562268ae32d416f9051efe60b869f14 : FAILURE in 44m 24s\n- tempest-full https://zuul.opendev.org/t/openstack/build/bd535885af2f4b1195e2eceb5c826474 : SUCCESS in 1h 26m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0b15067f2f984506bc25b69518f6396e : FAILURE in 11m 06s","accounts_in_message":[],"_revision_number":2},{"id":"f205917d83e37b231563d13cbf5e827134bf43d2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-09 15:41:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d94ebb7b278f19f85978785922b320f6e9209c37","revisions":{"44ef067a1e15a2a188e07adff389e04f3997b130":{"kind":"REWORK","_number":1,"created":"2020-12-09 09:55:24.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/70/766170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/766170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/766170/1"}}},"commit":{"parents":[{"commit":"453513c0f8619357a8fb1b624d898d4cff819eb0","subject":"update constraint for castellan to new release 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/453513c0f8619357a8fb1b624d898d4cff819eb0"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 09:53:21.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 09:54:53.000000000","tz":0},"subject":"Cap bandit at 1.6.2 when using python2","message":"Cap bandit at 1.6.2 when using python2\n\nThe 1.6.3 [1] release has dropped support for py2 so cap bandit to 1.6.2\nwhen using it.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n\nChange-Id: Ibdc34ab30314d4ed6c3133715d77486b92ac71b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44ef067a1e15a2a188e07adff389e04f3997b130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44ef067a1e15a2a188e07adff389e04f3997b130"}]},"branch":"refs/heads/stable/train"},"d94ebb7b278f19f85978785922b320f6e9209c37":{"kind":"REWORK","_number":2,"created":"2020-12-09 13:44:05.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/70/766170/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/766170/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/766170/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/766170/2"}}},"commit":{"parents":[{"commit":"453513c0f8619357a8fb1b624d898d4cff819eb0","subject":"update constraint for castellan to new release 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/453513c0f8619357a8fb1b624d898d4cff819eb0"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 09:53:21.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-09 13:42:47.000000000","tz":0},"subject":"Cap bandit at 1.6.2 when using py2","message":"Cap bandit at 1.6.2 when using py2\n\nThe 1.6.3 [1] release has dropped support for py2 [2] so we need to cap\nbandit to 1.6.2 when using py2. In order to avoid making this change in\nmany projects across stable/train and all prior releases this change\nremoves bandit from the blacklist and adds it to upper-constraints.\n\n[1] https://github.com/PyCQA/bandit/releases/tag/1.6.3\n[2] https://github.com/PyCQA/bandit/pull/615\n\nChange-Id: Ibdc34ab30314d4ed6c3133715d77486b92ac71b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d94ebb7b278f19f85978785922b320f6e9209c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d94ebb7b278f19f85978785922b320f6e9209c37"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
