)]}'
{"id":"openstack%2Frequirements~766426","triplet_id":"openstack%2Frequirements~stable%2Frocky~Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","project":"openstack/requirements","branch":"stable/rocky","topic":"ensure-virtualenv-stable/stein-stable/rocky","hashtags":[],"change_id":"Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","subject":"Ensure virtualenv is installed locally","status":"MERGED","created":"2020-12-10 15:09:06.000000000","updated":"2020-12-13 09:11:52.000000000","submitted":"2020-12-13 09:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766426-ensure-virtualenv-stable/stein-stable/rocky","cherry_pick_of_change":738644,"cherry_pick_of_patch_set":2,"meta_rev_id":"1af44619b45a1666723ae5aeb47d1329abf483d0","_number":766426,"virtual_id_number":766426,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-13 09:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-12-10 15:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-12-13 06:16:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-12-13 06:16:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 15:09:06.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-10 15:36:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-10 16:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 18:47:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2020-12-11 16:00:20.000000000","updated_by":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"reviewer":{"_account_id":12907,"name":"Jan Zerebecki","email":"jan.openstack@zerebecki.de","username":"jzerebecki"},"state":"CC"},{"updated":"2020-12-13 04:39:04.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-13 06:16:05.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"64d76e7aa3a315d9c52d233d6fe630f69aba0807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-10 15:09:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"cb3592dcb79e99d789c7f7a8778483406ad306c0","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-10 15:50:47.000000000","message":"Patch Set 1:\n\nI checked this using a horizon stable/rocky patch to cap bandit version\nhttps://review.opendev.org/c/openstack/horizon/+/766129 and the requirements-check started to work.","accounts_in_message":[],"_revision_number":1},{"id":"a423186b3836b99928e1a7bc0a0c4d2bd53be351","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-10 15:53:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2e9992f272369caa857ae33d5db6d0df5a43076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 16:10:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e18bac3497433b93efd0c87803a505 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34f0918ba27046aea34c34d67e2daaa5 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5fe3cd00a9cd4b23982bc5604c646465 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c0ab33a4eb1747349db9c9fc963992d7 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bebf374f55e48bb9807f6eb0c9d406b : SUCCESS in 4m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbb0b30d10784695ba7a905d5fee0cc2 : FAILURE in 20m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d71d4cf0b5934135947e5b4cb7510379 : FAILURE in 19m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3fa7abaa242044a6805e98c6e5219cd5 : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fda5080c2f0bcbc2733d14ce2ecb76471119aa8e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-10 18:14:59.000000000","message":"Patch Set 1:\n\ndevstack related jobs failed during installing bandit as swift dependencies [1], but a swift change on bandit in stable/rocky [2] is waiting for this backport. It is a deadlock.....\n\n[1] https://zuul.opendev.org/t/openstack/build/d71d4cf0b5934135947e5b4cb7510379/log/logs/devstacklog.txt#5423\n[2] https://review.opendev.org/c/openstack/swift/+/766491","accounts_in_message":[],"_revision_number":1},{"id":"13d23c449dfaac28011f4099b105108f31f33f27","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 18:47:06.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e devstack related jobs failed during installing bandit as swift dependencies [1], but a swift change on bandit in stable/rocky [2] is waiting for this backport. It is a deadlock.....\n\u003e \n\u003e [1] https://zuul.opendev.org/t/openstack/build/d71d4cf0b5934135947e5b4cb7510379/log/logs/devstacklog.txt#5423\n\u003e [2] https://review.opendev.org/c/openstack/swift/+/766491\n\n@Akihiro I think if we backport this patch further till stable/pike, and merge that first (the grenade is non-voting there), then hopefully we can merge everything branch-by-branch backward from there.","accounts_in_message":[],"_revision_number":1},{"id":"ad8880f72c5aabd72daf39780a9032aae7f2a271","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 18:56:00.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e devstack related jobs failed during installing bandit as swift dependencies [1], but a swift change on bandit in stable/rocky [2] is waiting for this backport. It is a deadlock.....\n\u003e \u003e \n\u003e \u003e [1] https://zuul.opendev.org/t/openstack/build/d71d4cf0b5934135947e5b4cb7510379/log/logs/devstacklog.txt#5423\n\u003e \u003e [2] https://review.opendev.org/c/openstack/swift/+/766491\n\u003e \n\u003e @Akihiro I think if we backport this patch further till stable/pike, and merge that first (the grenade is non-voting there), then hopefully we can merge everything branch-by-branch backward from there.\n\nI\u0027ve backported this patch till pike, let\u0027s see if we can merge there.","accounts_in_message":[],"_revision_number":1},{"id":"6714fdcfb7311f169387e7afd0e3710b14c02f84","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-12-11 01:42:58.000000000","message":"Patch Set 1:\n\n\u003e @Akihiro I think if we backport this patch further till stable/pike, and merge that first (the grenade is non-voting there), then hopefully we can merge everything branch-by-branch backward from there.\n\nPerhaps it is true for grenade jobs, but I think it is not true for the failure in the tempest job.\nThe tempest job does not refer to an older branch.","accounts_in_message":[],"_revision_number":1},{"id":"914031754404e57a2b64f7befdc76ef2f716ecdc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 04:39:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f54fb16556aff98fb6452f6a43d8cdc40e5a5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 06:02:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43fa8c57c3b14e4f8794b0f07422acb5 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1caa535ea1a944dba2590c850d2ce306 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/847f451a02e4457e9e2833863efbccb8 : SUCCESS in 4m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0ad0f5f511544e6dbc1aa31538e34eae : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/109d4591d2234eafafac82eeb20c8dfd : SUCCESS in 3m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a82ffc0c2444070926073640564e29d : SUCCESS in 46m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c96763ebcbe54646af433e16f64e474d : FAILURE in 1h 00m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/867420ae9e5d4b838eb651ad9c94684d : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":1},{"id":"199f2aa084ad261477efbcfc10255a788f2b90e3","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 06:16:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a7feb80508dc55389ddd796a0c94d37dffa34d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 07:51:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6599569a644f411b8fcd2527989821ef : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52fc084c7caf40928b5292b0cb667bbe : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d7ec7295f7724dc5bfc37c519074fbbe : SUCCESS in 5m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90e0877e6ab0454391e4a87141195f75 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26dc8aca6ba44c42a2c803fa131dd0c8 : SUCCESS in 7m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c39fad632f340d1815d6de95a367b33 : SUCCESS in 52m 21s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/186d03ad0cc6492bb24c218a1a0d766e : SUCCESS in 1h 18m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/49b0965e24ff4442a559383f33d1e254 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"07d9168f7fe3ff32e642213e7bac5a7fb4da8390","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 07:51:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e403578a29d4c136ef657cc75506640678b0aa02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 09:11:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/669cb780cb044ab89a06727c9864ef78 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/05c885eac827422587dac500ac56518e : SUCCESS in 3m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7163ef21743f415382b2479f0b5269cd : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/112d9608558242fabc38ddbe4f124478 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63dd475c817e4bdd8ed6cfdc9b847ca7 : SUCCESS in 4m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/38bf56a8e16e4e75be9f942cd2f42128 : SUCCESS in 1h 01m 56s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/05e4957da4af4228aef593bf754fa838 : SUCCESS in 1h 07m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/75933830e33942829df23bad05864979 : SUCCESS in 48m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1af44619b45a1666723ae5aeb47d1329abf483d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 09:11:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4e0d76a0c9888bbcfa973ca15673b8120971ae2","revisions":{"c4e0d76a0c9888bbcfa973ca15673b8120971ae2":{"kind":"REWORK","_number":1,"created":"2020-12-10 15:09:06.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/26/766426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/766426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/766426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/766426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/766426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/766426/1"}}},"commit":{"parents":[{"commit":"82b5132a0f80ff70b7af522c35d793b8c547b0d2","subject":"Merge \"Ensure pip is installed locally\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/82b5132a0f80ff70b7af522c35d793b8c547b0d2"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2020-12-10 15:09:06.000000000","tz":0},"subject":"Ensure virtualenv is installed locally","message":"Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n(cherry picked from commit 051549d82e354692de707d16a4bee361d69066c7)\n(cherry picked from commit 510875f4bd7a2112a18134e18c95579e7d2e0450)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4e0d76a0c9888bbcfa973ca15673b8120971ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c4e0d76a0c9888bbcfa973ca15673b8120971ae2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
