)]}'
{"id":"openstack%2Frequirements~766485","triplet_id":"openstack%2Frequirements~master~Ifc8e4b55bfc3f3de5a2eb98b6e9549b7b9d68cce","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ifc8e4b55bfc3f3de5a2eb98b6e9549b7b9d68cce","subject":"Add django-pymemcache","status":"MERGED","created":"2020-12-10 14:43:43.000000000","updated":"2020-12-10 18:16:18.000000000","submitted":"2020-12-10 18:15:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766485","meta_rev_id":"dfb456793ad46b7969e1c15026ae7db786f5d706","_number":766485,"virtual_id_number":766485,"owner":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-10 18:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-10 16:53:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-10 14:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-10 16:53:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 14:45:11.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-10 14:45:11.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-12-10 14:45:11.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2020-12-10 15:07:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-10 16:47:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 16:53:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c814fbc8d1ee0aa1a4a6c5e7b4b3d0b58beb1a27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2020-12-10 14:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38dc7cac5856ecc99546a2e4af13f1e7db5efdbe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-10 14:49:28.000000000","message":"Patch Set 1: Code-Review+2\n\nAs we already directly discussed about this together, I\u0027m fine to go ahead with these changes.","accounts_in_message":[],"_revision_number":1},{"id":"5e7ac670134c985fca54f72b573ef7c9391f8e19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 15:07:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/71472db8a477490dae0d820e5a2abd44 : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"93171a855ee38f9eca7ec1f0b29ed0c8eb7978d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 16:47:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e97dfa9368ce4d60ad69957bede8bf12 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f408f2b4244ec78da43d36563a002a : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/587714294b6c4011b596270d67771c2c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0227b97ae2c45ea897569d8ef03f724 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc658ed633f41c9a6644eddae21259d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e23d1674f3bd434bbdbe4badd8d6ff1f : SUCCESS in 5m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55da688a461d4eec93eccded72d599e7 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f9fadefc1a44fc9b46ad1236ac5c237 : SUCCESS in 16m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d29174de41f44a1cbf57354ffcef6f94 : SUCCESS in 13m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3e573be7e8c84a5f8d9726d9c67a9b1d : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/40e03801fabb4c68b9e96ec91d34ebfb : SUCCESS in 5m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e47c651b1e4548aa935c6720dbfb2966 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/776710dd8da24ecaa6d28649ab2f68a8 : SUCCESS in 6m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6fecd01ec1e84735b270493938454fb7 : SUCCESS in 6m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/36a76c3435e44bfb8b86529843772709 : SUCCESS in 14m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d40693bc9ed147ea96efff6ec10b91cf : SUCCESS in 26m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0fdc001d30c94395a544c51f3df32598 : SUCCESS in 13m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/070f18ea373144a6b4912486da37194f : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3d1c31072074384a26afa1fcdd98c2e : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c3bc03981d97401d94fdabc79dd9856d : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db9d116312c54ff69ef97f47b15deebe : SUCCESS in 6m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9d9f0bc8337b478587cfe60d70685666 : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd5da60b5d2b4dd897cd8cb6548f1d15 : SUCCESS in 1h 04m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c85d35db17b4436baf1960d731190e20 : SUCCESS in 1h 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"721f09780a79f70c4e746bd8a29b94b02c15be02","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-10 16:53:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5bd38e1bebaa03d60504584345ce5c344e15152","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 16:54:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5aace91f3e0cbc55a9acba172b2d886474334e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:15:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c686e3e58de7475a8e2346966c8ba474 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54eb55e5464d463295b7164c53920b91 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77bdc7d495dc432b816eca8980bb0265 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa2ca91777dd4856b3223b0604402761 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa8833b4a54e4a74aaf965ab27b1d8c9 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/06b21ece02d24457b9509e7cd53da2c7 : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00ab878fafdf41178805adfab45d2706 : SUCCESS in 8m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d4082acfad104d93ab9106d56ea264c1 : SUCCESS in 8m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7a481f97f08e4c35b7b64a9db6280dd9 : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6f59bf8c8c8648a28e5d67641a42a917 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90f04f3144094189887ee4e5fb6fed65 : SUCCESS in 13m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/57867067349a4ad1a5b7407e6305d4d3 : SUCCESS in 6m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c11c3a191f14fe69fc3cda46952f6be : SUCCESS in 6m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6aa7d9ce5ae40e6b247186308285dc7 : SUCCESS in 23m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5415488b7ee14eadbc0d2030b884c3f5 : SUCCESS in 29m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/170bff1f347145d280d8670ede263490 : SUCCESS in 14m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f0b53f5ce264f48bd0dc37a23f19ecb : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/67650bda826e48c7a4a0048d1027f3b0 : SUCCESS in 19m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b7a4d88960d1475782329134e01272e2 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2da2cc6055b34a52b2ff80ff4d91ea00 : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/49087af0b4e343d6a96d3a3b59abead2 : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa6562fc96ef4023883623157460c046 : SUCCESS in 48m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21f65193a54e4f09a45136f849fb3947 : SUCCESS in 1h 17m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6cfb5c72df03b6eab22c6667d25e8226cacd1b4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:15:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dfb456793ad46b7969e1c15026ae7db786f5d706","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 18:16:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c74213920f34230a6505b225c832554 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"707dde1fd4a6c52bf2e353d5b9334b226dc7c8f8","revisions":{"707dde1fd4a6c52bf2e353d5b9334b226dc7c8f8":{"kind":"REWORK","_number":1,"created":"2020-12-10 14:43:43.000000000","uploader":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"ref":"refs/changes/85/766485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/766485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/766485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/766485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/766485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/766485/1"}}},"commit":{"parents":[{"commit":"b4e2a19d7d1e7a717fe27d5b167d93ed3371e345","subject":"Merge \"update constraint for python-saharaclient to new release 3.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b4e2a19d7d1e7a717fe27d5b167d93ed3371e345"}]}],"author":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-12-10 14:17:51.000000000","tz":60},"committer":{"name":"Moisés Guimarães de Medeiros","email":"moguimar@redhat.com","date":"2020-12-10 14:17:51.000000000","tz":60},"subject":"Add django-pymemcache","message":"Add django-pymemcache\n\ndjango-pymemcache is needed by horizon in order to support TLS when\nusing Memcached as cache backend. While Django has its own in tree\nmemcached backends (django.core.cache.backends.memcached.MemcachedCache\nand django.core.cache.backends.memcached.PyLibMCCache), both\npython-memcached and pylibmc rely on libmemcached (C library) which\ndoes not have a maintainer right now and lack TLS support.\n\n* Is the library actively maintained?\n  Yes - there is a release from six months ago [1].\n\n* Is the library good code?\n  Yes. The codebase is quite small and is a wrapper around pymemcache,\n  which is already an OpenStack requirement.\n\n* Is the library python 3 compatible?\n  Yes, 3.6+\n\n* Is the library license compatible?\n  Yes, Apache-2.0.\n\n* Is the library already packaged in the distros we target?\n  No.\n\n* Is the function of this library already covered by other libraries\n  in global-requirements.txt?\n  No.\n\n* Is the library required for OpenStack project or related dev or\n  infrastructure setup?\n  Yes, it is a requirement to have TLS support for Memcached in Horizon.\n  We have a patch on for puppet-horizon [2] and another one is comming\n  for THT.\n\n[1] https://pypi.org/project/django-pymemcache/1.0.0/\n[2] https://review.opendev.org/c/openstack/puppet-horizon/+/765803\n\nChange-Id: Ifc8e4b55bfc3f3de5a2eb98b6e9549b7b9d68cce\nSigned-off-by: Moisés Guimarães de Medeiros \u003cmoguimar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707dde1fd4a6c52bf2e353d5b9334b226dc7c8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/707dde1fd4a6c52bf2e353d5b9334b226dc7c8f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
