)]}'
{"id":"openstack%2Frequirements~766530","triplet_id":"openstack%2Frequirements~stable%2Fqueens~Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","project":"openstack/requirements","branch":"stable/queens","topic":"bug/1907438","hashtags":[],"change_id":"Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","subject":"Ensure virtualenv is installed locally","status":"MERGED","created":"2020-12-10 18:51:25.000000000","updated":"2020-12-13 04:35:56.000000000","submitted":"2020-12-13 04:35:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766530-bug/1907438","meta_rev_id":"984301b0fd48e755ae29fc001925929fba2d9792","_number":766530,"virtual_id_number":766530,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-13 04:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-13 03:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-13 03:26:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 18:51:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-10 19:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-13 01:22:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2020-12-13 03:26:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a66cbd51bb79af20e7272a5d975f8d3fcfe3d751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 18:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b53fee266bd0bccac73b063d65704a01d30666b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:16:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf81723883446b7bd101c4593c7e408 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6496ff3a9294bbe906fadd792bf7f52 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/20195d8114ee4406ad3817b0a01015e5 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67b6e9543b1c4f70a21e9f6eec862482 : SUCCESS in 3m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/10974adcaac2437eaf2179965ecc4ea3 : FAILURE in 12m 39s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/8faedec97ccb41fe8bd93a48c3b046e6 : FAILURE in 14m 50s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b87bb8948799408db29d9525c5822292 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d6b7306039fb4475cc96ba862e81e4d2e3db4667","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 01:22:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac4ac8fb99a43ccec7e6d7fe42e99fa836a26686","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 01:22:31.000000000","message":"Patch Set 1:\n\npike merged","accounts_in_message":[],"_revision_number":1},{"id":"200ba702cef18f5218394a77a88e4009dfb48667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 02:55:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/011e7044e18148baae99e7a78b776789 : SUCCESS in 3m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09a321ac9bcf4877a9dbab3c3cbcfbeb : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ce1d80abe8624cf391fba2397969d8f0 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c4fc385c58a4b84b61f545d77630d8b : SUCCESS in 4m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7527ca3befc74db089b38c6bf024ae31 : SUCCESS in 1h 05m 27s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/fa30c4c868024a16b234950b637a50c4 : SUCCESS in 1h 27m 11s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d0191e65f8814ad1a684363ac10381bc : SUCCESS in 57m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c437cf4fc28662ecec0504acfa0b8b46e2a76139","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 03:26:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1624a0a3eea03d1fb168acf64a6c10f3b424e2bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 03:26:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0591b5669465f4dc6b7ae010f4195f2b8c5305cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 04:35:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cde69406cf4e55945c79aa248c22a4 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a62f6724ba194f448e1c99fc7e92798b : SUCCESS in 3m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/013b0a60d4214a4d882667ab5bf90dcf : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d198382d93f949cab63d44c8ea182170 : SUCCESS in 5m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b8ab1c752a44c9ca6fb194531b82884 : SUCCESS in 50m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c74b96923b5346ba83527d61ed0c6f7e : SUCCESS in 1h 08m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a535d44689b84e73a39094e3100d1860 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"984301b0fd48e755ae29fc001925929fba2d9792","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 04:35:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6e661b8cf2e2f75a303a0a2327b723238be506b","revisions":{"a6e661b8cf2e2f75a303a0a2327b723238be506b":{"kind":"REWORK","_number":1,"created":"2020-12-10 18:51:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/30/766530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/766530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/766530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/766530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/766530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/766530/1"}}},"commit":{"parents":[{"commit":"5cdb94c26bf26b085302517f409eac1d448f709f","subject":"Revert \"[stable-only] Upper constrain wheel to 0.34.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5cdb94c26bf26b085302517f409eac1d448f709f"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-12-10 18:50:19.000000000","tz":60},"subject":"Ensure virtualenv is installed locally","message":"Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n(cherry picked from commit 051549d82e354692de707d16a4bee361d69066c7)\n(cherry picked from commit 510875f4bd7a2112a18134e18c95579e7d2e0450)\n(cherry picked from commit c4e0d76a0c9888bbcfa973ca15673b8120971ae2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6e661b8cf2e2f75a303a0a2327b723238be506b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a6e661b8cf2e2f75a303a0a2327b723238be506b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
