)]}'
{"id":"openstack%2Frequirements~766532","triplet_id":"openstack%2Frequirements~stable%2Fpike~Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","project":"openstack/requirements","branch":"stable/pike","topic":"bug/1907438","hashtags":[],"change_id":"Iac5e49a58a46d9c33632ae0ca4096fc102fe7123","subject":"Ensure virtualenv and pip and unblock gate","status":"MERGED","created":"2020-12-10 18:55:25.000000000","updated":"2020-12-13 01:03:49.000000000","submitted":"2020-12-13 01:03:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766532-bug/1907438","meta_rev_id":"1c8a0fdbd612def6f2cb9d28610416959b258beb","_number":766532,"virtual_id_number":766532,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-13 01:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-11 01:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-13 00:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-10 18:55:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-12-10 19:06:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-10 20:15:14.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"CC"},{"updated":"2020-12-10 20:17:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-12-10 20:17:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-12-10 20:18:31.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f535edeedaf93616fd1c6f2f1949b69b6939929c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 18:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d0b7da6493b6a5bbd4fbb3f9c49e2a4ae4e14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:06:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/289eea23083a4a4b8675fa0416a4b3dd : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79c248704e4a41f1aeded0a4b4e29b8e : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4568e99c611f4ff7ae16293e9e33b05b : FAILURE in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/039e5432264c4fc99eb480a460c460d9 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"466a5fa47db6fbe6aa792dc9d9c13074a20a04f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 19:49:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90fa407e9ce162152fc654f36686b528e6c2d5cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-10 19:59:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75ffd5245f454f6596da3c069dbb2027 : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab15dc6458fd4f28b43335a01576d72d : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bdd4e62580b6432286f0f116840bf690 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0463cca00d694d8181b61733f75b8529 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b749ad4ec2cbb5cfbc9075476b93a815c703acd6","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 20:15:14.000000000","message":"Assignee added: Akihiro Motoki \u003camotoki@gmail.com\u003e","accounts_in_message":[],"_revision_number":2},{"id":"2b3de5d5c4cbd3f9c32bee30a6bd8e10cf52d3c1","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-10 20:15:26.000000000","message":"Assignee deleted: Akihiro Motoki \u003camotoki@gmail.com\u003e","accounts_in_message":[],"_revision_number":2},{"id":"4fe489a127989616b56cde467e73ecf228745b92","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-11 01:23:49.000000000","message":"Patch Set 2: Code-Review+2\n\niirc we need to work backwards","accounts_in_message":[],"_revision_number":2},{"id":"50da7c3f42ad75fb5a955304419abb921a1a4c3b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-13 00:53:54.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f07588df28642054b07c912f05beb0d3fc09510b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 00:54:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1c8f5edebeab3af1c70ca812c5161d04c4b7b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 01:03:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ece0f13f9b469ab81af10f79ab603b : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0091fb6617334b08bd521213956e011f : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d0d6e0ff44a40a7abd08a2a0f57b298 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aad9be0c4bb64727a2ba9bcd309e4de7 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1c8a0fdbd612def6f2cb9d28610416959b258beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-13 01:03:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5402f95ddc9155fd603d9fb2e2fc089b3658035","revisions":{"3c958e85a224f1fef5b0418b22b4ab993582e3d4":{"kind":"REWORK","_number":1,"created":"2020-12-10 18:55:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/32/766532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/766532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/766532/1"}}},"commit":{"parents":[{"commit":"384a9478d007e3674114de5d3e4979a08aa73f34","subject":"Remove legacy-tempest-dsvm-neutron-full","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/384a9478d007e3674114de5d3e4979a08aa73f34"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2020-06-29 21:43:56.000000000","tz":-240},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-12-10 18:53:45.000000000","tz":60},"subject":"Ensure virtualenv is installed locally","message":"Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nConflicts:\n  playbooks/requirements-check.yaml\n\nNote(elod.illes): conflict is due to missing patch about ensuring pip\nlocal installation (Id4969156383545eb94462677e3321b89c1160b48).\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n(cherry picked from commit 051549d82e354692de707d16a4bee361d69066c7)\n(cherry picked from commit 510875f4bd7a2112a18134e18c95579e7d2e0450)\n(cherry picked from commit c4e0d76a0c9888bbcfa973ca15673b8120971ae2)\n(cherry picked from commit a6e661b8cf2e2f75a303a0a2327b723238be506b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c958e85a224f1fef5b0418b22b4ab993582e3d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c958e85a224f1fef5b0418b22b4ab993582e3d4"}]},"branch":"refs/heads/stable/pike"},"e5402f95ddc9155fd603d9fb2e2fc089b3658035":{"kind":"REWORK","_number":2,"created":"2020-12-10 19:49:59.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/32/766532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/766532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/766532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/766532/2"}}},"commit":{"parents":[{"commit":"384a9478d007e3674114de5d3e4979a08aa73f34","subject":"Remove legacy-tempest-dsvm-neutron-full","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/384a9478d007e3674114de5d3e4979a08aa73f34"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-06-17 21:19:06.000000000","tz":-300},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-12-10 19:48:01.000000000","tz":60},"subject":"Ensure virtualenv and pip and unblock gate","message":"Ensure virtualenv and pip and unblock gate\n\nThis is a combination of 2 commits.\n\n1. Ensure pip is installed locally\n\nRecent changes to the base nodepool images removed the system-wide\ninstallation of pip and virtualenv. As a result of this, some\nrequirements-check jobs are failing with an error that pip could not be\nfound.\n\nThis adds the ensure-pip role as part of pre_tasks in our\nrequirements-check playbook.\n\nAlso necessary to squash:\n\nFix failure with setuptools \u003e\u003d 49.0.0\n(cherry-picked from commit d30e19237f8eb8be9e1a736f1d4503c8cfcd74b2)\n\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n(cherry picked from commit 74a5bed896893ecd3bdedb100439f773af7c7f5e)\n(cherry picked from commit d303fe350d74d2d40aa5a591ee9d372a529d10be)\n(cherry picked from commit d43e0a585719930516fd82504d84e3867512c7f6)\n(cherry picked from commit 0f51382318eff21f9d5a02375bd55b3f805fcc62)\n(cherry picked from commit aa3018d58e739951b1cc6697e2a207a73739976f)\n(cherry picked from commit bc7833bb37d89a0973cef88621ba7ade3be574a1)\n\n2. Ensure virtualenv is installed locally\n\nThe stable branches that still support py2.7 need to use\nvirtualenv, so make sure it is available.\n\nChange-Id: Iac5e49a58a46d9c33632ae0ca4096fc102fe7123\n(cherry picked from commit 051549d82e354692de707d16a4bee361d69066c7)\n(cherry picked from commit 510875f4bd7a2112a18134e18c95579e7d2e0450)\n(cherry picked from commit c4e0d76a0c9888bbcfa973ca15673b8120971ae2)\n(cherry picked from commit a6e661b8cf2e2f75a303a0a2327b723238be506b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5402f95ddc9155fd603d9fb2e2fc089b3658035"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5402f95ddc9155fd603d9fb2e2fc089b3658035"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
