)]}'
{"id":"openstack%2Frequirements~767031","triplet_id":"openstack%2Frequirements~master~I729ea78158fa49fc641deb93517d314bc4f13ee5","project":"openstack/requirements","branch":"master","topic":"retire-karbor","hashtags":[],"change_id":"I729ea78158fa49fc641deb93517d314bc4f13ee5","subject":"Remove Karbor deliverables from Reqs","status":"MERGED","created":"2020-12-14 23:35:36.000000000","updated":"2020-12-16 18:09:06.000000000","submitted":"2020-12-16 18:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767031-retire-karbor","meta_rev_id":"8808382a519895f20e7a0afb2b06eb1cc45d626f","_number":767031,"virtual_id_number":767031,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 18:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-15 02:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-16 16:27:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-12-16 16:27:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 00:01:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-15 01:19:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-15 02:09:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-16 16:27:40.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"7cdf519df995a79710308bc4aa1d20d0bccabfc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-12-14 23:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"146c19e457b1d799b79b9b5e43dc6b84e727784f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 00:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5d2cf49b922d42da9761184b3d90b029 : SUCCESS in 20m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b2ad23db89fbef4a18b231000bb877b8914b644b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 01:19:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb0dc1438a247c385d2a36cec330e31 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13325853343d48768e828db37841ad19 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d8723dcd9b54905aa9d6b112076da96 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e0a6775d0ee4a419b552b3e04904ba0 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cab65bd67ca44949b94ece04f2f4a7c : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60aec2c5fdee4899a3762594be009baa : SUCCESS in 5m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/60b9e2db2a5046cb8a5d8b629d2161ee : SUCCESS in 5m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fde7ff313f904c1b9638d00a0478a5af : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b135ff26fdec4a5eb74742df6358c2aa : SUCCESS in 9m 29s\n- requirements-tox-validate-projects https://zuul.opendev.org/t/openstack/build/7feffb7dc5454e02ab2b9fd2b22243cf : FAILURE in 4m 03s (non-voting)\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/347a902f334e45af9cd6a67625b4e1ea : SUCCESS in 8m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43b2ab4176b1408485417ef3f36060ee : SUCCESS in 4m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/77a0e2cf989e454d8c4bfad56a972f2b : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cdeb9f78cfac4ca69bf23fbd55468fdb : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/023f3deeeeeb408c84279e8f624e100b : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bfb411d6f6ef4ba8b60be1f73c3a20db : SUCCESS in 14m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d67d58cd8e084995ab51edc745d61e99 : SUCCESS in 26m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7134861cedc54fe7ac2d5fc95604bd6d : SUCCESS in 14m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba834951c1c849a5849c4ed99ab9054c : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/092481411faa4c14afe861653b3f7858 : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3c1824594ee0448aac035f692a95e3c7 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a08008624ab54ac2a830b090617c62fa : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc7f8959847b4d0e9cf295ea2a91d814 : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf5693eb1f774757be3b6ddef1c467f3 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dad85f07739545799f882723a1ee3510 : SUCCESS in 1h 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e2a35e7e8d2b84be95ccbf081c9f2e2e749a5602","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-15 02:09:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cc4fdb90c284e641914f9ce893ebd8d81e83714","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-16 16:27:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c0cb6dc53eb7e4c346312248566d65ba670d7e40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 16:28:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfda84f983bf67a630c1d7116a92691f8e0689d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:07:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ae63749e8a4efea9adba28d2360e70 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7ddc0add67e48fdbe8fb5b683a9c0d5 : SUCCESS in 9m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c1411abf1f547efb532856dff5d5094 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c99f60593044ad1af30f17c51afba0a : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c04ec6410ac94d91956555fa2e34a46f : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69001104e9404a48b427154751fd2af4 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/180438d3d56044d9ba3fb4a5a99ca283 : SUCCESS in 9m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3f6394f0e62462884e12fe5c86b20f8 : SUCCESS in 9m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/808edd0eea984ef7ad4d674caba3d479 : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77b4379caf01441ab4b640d682504053 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f2be9fffa0f745bc96e30238f495b36c : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4124a1a34c684c8f8e89de2666789a1f : SUCCESS in 6m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eb524d8567534260a59181a0a80c7a45 : SUCCESS in 8m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e6ad9488779a4e91b0be36add4bde2fe : SUCCESS in 17m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9f69bb1a0ebe4498bc4e7f5bf8f7f5ae : SUCCESS in 25m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9bab4333588d4cdeb4fe99349bc1039f : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6e89321ca567427a88aecdad319603ec : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/18fb76cdaf23457db290cc5939cabdc8 : SUCCESS in 12m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d6a8898d308449c1843084e31205c2d1 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c2f3e8b41a844048c2cafe977526872 : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e1b79873e674472fbe156838f383a4a1 : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd9371827406473cb6c8aff33e571670 : SUCCESS in 54m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/671a8e8faab444e695523592acb29d5a : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"526fca65d5eb3a47c9231980e328b3948bca0ce8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:07:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8808382a519895f20e7a0afb2b06eb1cc45d626f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:09:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0162152486f346518b42907226b869b9 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b12c8794e87d97652981f6843c5cdfc989128040","revisions":{"b12c8794e87d97652981f6843c5cdfc989128040":{"kind":"REWORK","_number":1,"created":"2020-12-14 23:35:36.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/31/767031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/767031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/767031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/767031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/767031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/767031/1"}}},"commit":{"parents":[{"commit":"42e95166ffe584a722697048c6a674be03caad02","subject":"Merge \"Exclude packaging 20.5-20.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42e95166ffe584a722697048c6a674be03caad02"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-14 23:02:23.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-12-14 23:06:26.000000000","tz":-480},"subject":"Remove Karbor deliverables from Reqs","message":"Remove Karbor deliverables from Reqs\n\nKarbor is being retired this cycle so it makes sense to\nstop syncing its deliverables requirements.\n\nChange-Id: I729ea78158fa49fc641deb93517d314bc4f13ee5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12c8794e87d97652981f6843c5cdfc989128040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12c8794e87d97652981f6843c5cdfc989128040"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
