)]}'
{"id":"openstack%2Frequirements~767280","triplet_id":"openstack%2Frequirements~master~Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-12-16 07:13:20.000000000","updated":"2020-12-26 00:03:36.000000000","submitted":"2020-12-26 00:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767280-openstack/requirements/constraints/noclob","meta_rev_id":"2725f49c2701cc870fcef240198b23a9c25f0134","_number":767280,"virtual_id_number":767280,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-26 00:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-25 21:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-25 21:23:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 07:34:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 20:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-20 21:23:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"965e69a9bac85bd0a7ac86152f19ca8ebf78ba85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-16 07:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbb53f26c436a128b9b5cd6492853d31dac26601","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 07:34:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/146ecc2dd92c46fc9fbd964a859f6f1e : FAILURE in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6ca47096ec798fa535d18da8f08884a0f5484938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 20:44:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abad92260ca9421b985ea87098edacfc : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e068be697694469803235a615ca940f : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb1cb33d345e4bd994ae2843dfdd5720 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f8d07ab5a9d4812a333fe7e3276f8ec : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/185b8f1cac1c4a66bca145e65c8aedbe : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4c80a34b2414a5d888b7eea8d819815 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0e8992777104c33afb6910e3e607fcd : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/671231106957488c92367d59dd3014b5 : FAILURE in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6dd479f441bc4025970d8a6a59db2867 : FAILURE in 5m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9c714e372588443890f6f8d7d2aeb1f7 : SUCCESS in 8m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9f5280ede6104f8cb140e5b7d4d4fd0a : SUCCESS in 5m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bcb706dc4525462aaedaec42a4077d61 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c29bd52f00274b759bd839534aa387d5 : SUCCESS in 7m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6108d18afd44555885698ba2df58353 : SUCCESS in 6m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ed9fbf7abc8744c3bce613cd9ef058c2 : FAILURE in 14m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7b9bd8a6232e45f5be53746e65bc027e : SUCCESS in 26m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2e2c3193e6e0446abf3fa5a116b260ec : FAILURE in 11m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c146881a340f4f2bb444778282b59376 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bf53ba410be74bcabc3755a46ba5c7ec : SUCCESS in 19m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d7b7ee7bd504b48952afbb950cce53e : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1626c5914f1b4a72bd34befb3487eff3 : SUCCESS in 6m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4fc82e5a168a42bea4a6c25a0fb8e2c5 : FAILURE in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0013ac40af6c43918deee7e73f1f2f13 : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f5b24b4d2aa4deda43c81ce85894afd : FAILURE in 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c0133ab125823a895cd9c0fca4b3a62a6e3fee50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-17 06:24:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ed143f1039ee929f1a3a74180d41c21135aa0ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 06:40:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a9356525da64d9f8eebeec26d1c224f : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":2},{"id":"bb5473f751a6a648ca2d31420c5e770b789046b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 08:35:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efc495da7ff4899b84f26f6c3e7b049 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6d36c789c74e5ba3657545e681b0ec : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8a28591367643b9bd3578e4148d4c37 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69f35da3b45942238631113e4fdcf990 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/907d11c991bd44f1be2fc02d0dccdea6 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9881268390847428022cadd4901909e : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16fb1dac1a9f4114922fac6ce9a0cdbe : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/11b5b0cbf3654c37aa83b1beb8382946 : FAILURE in 3m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/08a1acf7cc6e48b5920ee3ae52e838e7 : FAILURE in 3m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d113e70ba3d548e480c279c41ce4a1f5 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f54fa3e99e9a489788886631fc684412 : SUCCESS in 6m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/31309264f45b4d80a9b69d7c22610429 : SUCCESS in 9m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af353dcaddf049ef95f9d699146f8a1f : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cd747c4fb5c2447f8569b484fe4138db : SUCCESS in 7m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/729fd7deef30464eb2af24ccb6b64c22 : FAILURE in 16m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/139cb7e3725d4885b1c32393fa7e9d8a : SUCCESS in 18m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a5edfe4bb02442e58a1458a1f37914c5 : FAILURE in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/608e27195c48499da73df47355d40c7f : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/09fb3261d02a43b28da2960dcbb807b3 : FAILURE in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b8dcfd9c920f4501a3d27d672b96ce5e : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecee8b791ca94c6581785ff361e89533 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ab4b9f616d040988153008097a56505 : FAILURE in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7116bbef247045d6bcdeab6b9574062c : SUCCESS in 57m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82def7d2ef3345ab81b6440b020e77e6 : FAILURE in 23m 42s","accounts_in_message":[],"_revision_number":2},{"id":"b4fdf8b80569774b05822f88ec670b7ed0f98054","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-18 06:26:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"079873713626c76dc35d8a7aab0b4cb49ac64ccd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 06:44:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64b84e595b3140d4860bb916221331d1 : FAILURE in 12m 48s","accounts_in_message":[],"_revision_number":3},{"id":"62c1991e12abdb88bcab6edc62f0283f9a79de52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 08:37:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e195af06014b1eb3f8e4b9f7853bcd : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bac52b590ff4cc3b2875f96a4c62ba4 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c7e36dc26ae473f97f6f30f0290c1fa : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/072dc2f2dabe499498659a00623001d1 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/175e9cb73ec642c9abdb92828b91491c : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d2825ec330c4c57bae5577fed420e80 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad411e5aaae7442396d8b8deb623c54e : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4e4eeb068e84725bf280205c5165ede : FAILURE in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9b99ab7b530f4ffe96732f2a8f22e653 : FAILURE in 4m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4f35002a6b404eb9b2c6e3a224ca4700 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b47dbf49602b44ffb42cc2f8af3f8bf9 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/464efe35d15646cdb53fce35c859a4f9 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/729d0724c0c24fa0a5c1ca6d8e3649cb : SUCCESS in 6m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/777762546b9f43888ba9cc8cae60df3e : SUCCESS in 6m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e66bf6f7c59042908cf95f4fcf1d2fec : FAILURE in 14m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/11c3e29a8fb849d1907b92b6428f3c16 : SUCCESS in 24m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/283f36103a2742e78c548e99b2b834a6 : FAILURE in 12m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/96c2f17ce6c444a1abc860dbc3b365bb : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d86d43762da34852abeec4b98ab7eb06 : FAILURE in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ec56820f76c1417eb994874630d30c28 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99c75702b0e4440ca87f70d4e76b2059 : SUCCESS in 6m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/225dd7d8672248268cd61a48dd662990 : FAILURE in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d073437a836040ce9adae1ce0a20e3aa : SUCCESS in 1h 00m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56e35fb53407415399dae2021a05836c : FAILURE in 18m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6359cec7cb230f34cd8aa2bfc60466280351510d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-19 06:23:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d1a505d3a282bf4032e058083e42a19cd758de55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 06:56:42.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/889d7b92dce0480d8a466aca4f19dfb6 : RETRY_LIMIT in 1m 43s","accounts_in_message":[],"_revision_number":4},{"id":"c60c3e1a28a66c68367a7a0f36e94a41715d7d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-19 07:46:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b65cab4e9348df91958002fc4a2879 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08468f8e52544fe5b8bda61fc20d9fd5 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4880a820a3bf491e98170d87193af706 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/111252012a13416e88b3240df69e70a7 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75bf024fa4c04e5d93af56395a6de80f : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/581e88545a7140e39735c70944c52cef : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/557c3cdc5f6b4a3080d240b1b26717c3 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/862130ea28a44d02b86755a9d30b2a9b : FAILURE in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4b4e0a568d1e4dea8a7523a705262d3d : FAILURE in 3m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/983d6d83c65e440d9729c07b6f6a3867 : SUCCESS in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7ee2743424c24c6a8b6a8be98c8e5d4a : SUCCESS in 4m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/adb1aaddd969421c90d49f8fb8dd3cbd : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfabc456b3034e4981de98c6b81aa354 : SUCCESS in 7m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64041e4a271746eb80fea92949a754c2 : SUCCESS in 7m 25s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/426115d74312440d89be6d8a64b125b3 : FAILURE in 15m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/55d650214e4c46fe9ec2cacd3e10211b : SUCCESS in 25m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a601d76c4a44ebbac6f29dbc08a205a : FAILURE in 12m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b0f3a88d6d984305ab819b918ad1ad01 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fd93c7536da047ccb65eb1c26ec0aeea : FAILURE in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ec567fd4848640e181c81e6e810d094b : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/df7b2852fe284508a6a8c1a8928b2105 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aae1951cc0fd4a5cabf8e7bcf0626dd6 : FAILURE in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9309cee442864202b4609f68ce83da25 : SUCCESS in 57m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bbb76134da247cf87733df1dc744619 : FAILURE in 17m 06s","accounts_in_message":[],"_revision_number":4},{"id":"3fcaea54e5cadd2606ed2217c01e2cb35790a6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-20 06:17:47.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0836118f490caa6491fbda853653385347f053d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 06:32:45.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c4c22accab6d4bcdba3cbd8730c3c06f : FAILURE in 10m 25s","accounts_in_message":[],"_revision_number":5},{"id":"89ff07049cdf3664fac5259ed7983c1472e73503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 07:23:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406c93e986a84bc8a810e3d751644be6 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2121f42d3a3429a8024376e97f028f8 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62cbb5c460264bc79604abec626815ce : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5bc3f9a93ec44c3adcac6aa019332b0 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd7b3f3747454085a733fafe234a0a3c : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/734ae75643634fafa16362727d549834 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cf09db07bf44d9fbaece49024b0fd55 : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e57418ed027c41e7827a2f67c8ca548f : FAILURE in 5m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e1e5f094e6894d0ca9faca81ab158d6a : FAILURE in 3m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db5467e5cf7e4090acb39f27c52363d5 : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dfbf8a24fcd049dbb7cd0ed2cb04bdd5 : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dc6231d192c84adab14e4d8118ea5c91 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62f50865bccb4448a35eb01e796bbe25 : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0bfedf8b4c14942a0e090888b18fd9a : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bbd47e8e1e1242acb2db2543f25106f3 : FAILURE in 13m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a83c9044296d4073ba9c673d20d465a6 : SUCCESS in 27m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1d1fe28f7ed546be931d9f62205a0bfb : FAILURE in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ef8256303bab454aab9ec2e8b91bcba2 : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e03d63255d184cb19611cd7fc499d923 : SUCCESS in 21m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77771bf4e2eb4499ad0234a623e7ba79 : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d139cad44fa4e19b256588cbd5a9529 : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/15343291c2f0471b92c2ec56410deff3 : FAILURE in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32374cff84de448ab1dc11a22a8021f6 : SUCCESS in 49m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bab5e0aff3947beb49b0174ba05a047 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":5},{"id":"ad238edbfde022f3135c4c81bbfeb0fa2d975a25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-20 21:23:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4a2f3b137f831c5663b8eeda325994ddaa05e9b0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-20 21:24:17.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":6},{"id":"ddcd3dd9018aee06239d55bf12ec3597666c37bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-20 21:28:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c68e015cfdeb9111e73ea4bc6fa060b04d8a6903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 21:37:48.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/283252aa09f5498b9a3a8a8bafd8303c : FAILURE in 6m 18s","accounts_in_message":[],"_revision_number":7},{"id":"2ab5dd5b2d68a69031df1f337069570c301a0142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-20 21:40:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c315baf0064c27a5438bceee557098 : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cd6ba73c044b9187e2bc58b3335b62 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eef0d594f777480b988b550c22dbd494 : FAILURE in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0dd24a40e074d8ea844b7c3cc901a7e : FAILURE in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81432e5556004abb8795c5cd1465ef5f : FAILURE in 3m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/873d2938dbbe4325a0bce67c53b839ab : FAILURE in 2m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1a069ef0f964f48acb3830214ce4048 : SUCCESS in 2m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5b36c19a474a44eca3d48d30ed8d6935 : FAILURE in 5m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/28cd9f95d8014af7a9f4d7160ac8788a : FAILURE in 3m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bb59da93a78b46e4a8afc2a6664b0a41 : FAILURE in 4m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/00a627a167324b45bfacea9120e0feef : FAILURE in 3m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1f233e7280794a52b25f0e168808f3f0 : FAILURE in 2m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/958a38e9cec0495490afa2f375555c76 : FAILURE in 5m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/db34065e0c084ff388358542c3bfc1c0 : FAILURE in 5m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/27bdf026823f49419dd620ae47c24548 : FAILURE in 4m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dfcf43b0436c49518c6d0c079a95b5d0 : FAILURE in 3m 36s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ba3a2efe4cbe40719315aea396e3b54e : FAILURE in 4m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5bf561d24e824fae969a88192347f9b5 : FAILURE in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa62cbece79347bcb81d82af2971a741 : FAILURE in 3m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/15baba920c204e128672bc2eda92d3ff : FAILURE in 2m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13e8614bc0104e9d9fd9e7a14b4cef0f : FAILURE in 3m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/08d29f05fd7c43a4a29c40a663817862 : FAILURE in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e708434d63c54589836f99f542f81940 : POST_FAILURE in 10m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/814c2aaac6394d7aa7327554266da9e7 : FAILURE in 9m 13s","accounts_in_message":[],"_revision_number":7},{"id":"2d409a8afe7ca2528614e8da74d9eb1ae82b6524","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-21 02:23:35.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b1d8d2eea4d00db7559730c877b42bf2d84a33f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 02:33:41.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e7c1b2f601e049269063401d5366b1f4 : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":7},{"id":"135280d7467be563fb734ebbaee283f330075f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 02:40:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55d0794844446e2bd2518281a73d401 : FAILURE in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797cecae2acc4abea21bc4aca802908f : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5072ac27caf453f8dcb8e160c3b86f9 : FAILURE in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92f6c58fc1ba4a808903793c66953b01 : FAILURE in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eb3b8b0a6134776b0aa5ee9c347b347 : FAILURE in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4534adfd0614fa9964aad57052435ca : FAILURE in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7857ba42ec8494988689fa009375006 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af79213da1e1405fa2425961b27e4c91 : FAILURE in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fefbc46ebdeb400887310f8cc9666b37 : FAILURE in 4m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/10b9ad313f8e410e8d601022ab9d44db : FAILURE in 3m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b678c7d01f24534ac30b6317a1e9c8c : FAILURE in 3m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b2b0e28a737245989481ab30ecb136f6 : FAILURE in 2m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/072cf05f18c54f7e877337ccd1973ac6 : FAILURE in 4m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3ad0eeb522ed498489309387650b72bb : FAILURE in 4m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8573c4aa43f6445c816232ac00e50402 : FAILURE in 3m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f61d94072aa74d3299c87d126b6e5db7 : FAILURE in 4m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c1f2ce5ad5474bd9b5e5583a66e64353 : FAILURE in 4m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e37d881b2bba4869aa0fc3410d84cce0 : FAILURE in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f188ca2e9dc42dcaef01c23985c11b2 : FAILURE in 3m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b19e31becaf24b32a10d9e7712295fc5 : FAILURE in 3m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f10afd6591643b2b8c0e980c75cc4fd : FAILURE in 3m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b5e3f3da43d841f6b2762074042d9f54 : FAILURE in 2m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df509e5549554fcfa190f8c896fe8ff7 : POST_FAILURE in 14m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1f7678c4aff40b391e4ec660236f2d5 : FAILURE in 8m 14s","accounts_in_message":[],"_revision_number":7},{"id":"a489fb0f16d6fba30f599be98d6968d9f1ec5210","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-21 18:16:09.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"9b82c4ad21894948d05d2cc63abec0d19c413921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:26:33.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20646ab777e7408182323bd58470ea93 : FAILURE in 6m 18s","accounts_in_message":[],"_revision_number":7},{"id":"63a3f1fce2d124dacd7396ca870c375bdf0c1f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 18:39:36.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780f76032a3c4c81ab681399eb2035e4 : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d29a9aa80d604fe9b1bf85c808ee3f66 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834accd4b1474add92636fa72341d8e4 : FAILURE in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe093e48262429f85682012ed862a4a : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55e162d77f9431ba565e74a68516634 : FAILURE in 3m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/218aa603f7674e0481aa5bc38bed09f1 : FAILURE in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d52a000375d34805b4d5b6b8993bb1dd : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdedb7719b33467299c52b6a3ea7041f : FAILURE in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fd7f86925e514dd8b97277c8b6d2ad10 : FAILURE in 4m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d3d75d849fba4dc7a86b4f02d2d408fe : FAILURE in 4m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a28aa7c17cf44413aa5b7cfae6226831 : FAILURE in 3m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cb88a6c35b0f449590bd551334ac491d : FAILURE in 2m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/56c38134a09e4e249d4191624ffdd945 : FAILURE in 5m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/19e4f1d4491a4c7b90aa044485d69d62 : FAILURE in 4m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2d66cbc7ff4d4512847095701dff0aa3 : FAILURE in 3m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/16942e577719448798294ade04487ddf : FAILURE in 3m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9e8c6ed3f1704cd0ba1c6ec705eebb24 : FAILURE in 4m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f01687c00b543f095365f69fae7f711 : FAILURE in 2m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f4da1669cea4eb78bd20b132cc7f948 : FAILURE in 4m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2e19124000bd4be5b3b0396dfe24f164 : FAILURE in 2m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a532bf5aa7134afbaded2951c904072f : FAILURE in 3m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/72d7101e59894836a3dac9cfeb008fec : FAILURE in 3m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27c77cb2ad0f404b82c2151382a9670d : POST_FAILURE in 19m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/698674c5c1a44b4f8b14a1289f053322 : FAILURE in 13m 38s","accounts_in_message":[],"_revision_number":7},{"id":"62bdeb9543898775becb8a7a5885cbbbf68e7225","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-21 19:20:15.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"800a358a3bf202c3c533e1bca24736c60f901826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 19:36:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a315d2b13b2147bba86873584136a287 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":7},{"id":"086b18c469f3de0a86c79a6111c2f0fd5908db86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-21 20:09:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7ae5b8ee9140bc89e0ca6328ff58c6 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69d1f5d380d44dc9db34adb04c2c629 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06658dec100d4ce1aaf6a1de7dfa1bb4 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07324f87ab8f4e80ac75150215257044 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbd405228e364e099ee8890476cbbaa0 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f64e2778e24400fbe02141eb694e807 : SUCCESS in 3m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13525853cf8e40a9a33017c2d3adced0 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0afcc43d0f9441419c1120835b20f119 : SUCCESS in 7m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bc104dc9f6d343849bdd9f4e74dba97c : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/362c8e0e354c4f8693fc445491c4982a : SUCCESS in 8m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/161af2e2354e4386bd372ea939c87d91 : SUCCESS in 4m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d904c60f536040329211dc0b93175b26 : SUCCESS in 8m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7be1a944133c48929e40f692933244a2 : SUCCESS in 6m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3bf0444f22e349138fdb44531cfd871c : SUCCESS in 6m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/71291530e9b54fe0b26af6f705570033 : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1bd8a86ee7eb48128da773405503e2d0 : SUCCESS in 20m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/97aa5a3e022f44b5a71bffbb49ed0332 : FAILURE in 12m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b9feb5a7525449eeab67669419b59207 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/55e6e68d0cdc4ff4a35cdcfea8711cd9 : FAILURE in 15m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18975f071e804a329b98ed875faf0e1a : SUCCESS in 4m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc5386f2d5d346c2b63027bbea1435c1 : SUCCESS in 11m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e836a8962be743c38af260a57122aa7e : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e83d9bfddec74aa280ca8ec1646fe4a1 : SUCCESS in 46m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c21eab403c24520b46b0e7a4110d15b : FAILURE in 14m 47s","accounts_in_message":[],"_revision_number":7},{"id":"281fb9c4aa1e721d7d07495a888d564f67a07bf6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-22 20:19:10.000000000","message":"Patch Set 7:\n\nrecheck\n\nlets see if nova actualy fails or it was transient","accounts_in_message":[],"_revision_number":7},{"id":"618ecb15eb37baa62d1662417eda22e8c907a240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 20:52:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/117903fce5d74b9b8020461495b44394 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":7},{"id":"1eab71d6fbd91e575e49a6d1bfd0e6db76b4cdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 21:17:44.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc5dd169cfe430a8688c1e6472b4e71 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b91d5ab6560406c88ee14e67105fdc5 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fa3396175ce470b8c9624eb4fff28c3 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f98ea4f7f0941359f2fa5beab5f4c40 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867eca02dfca45c698907cee216c46d3 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6fb5628ef3b4f6da98599b921d80d57 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c336237c7d344449db563373a2b0dcd : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/018f24d1656b47c881a9f41049f73406 : SUCCESS in 14m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ebc21bea89e4ad0991f6424c4367aa6 : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3dc8b85099804e3499d58d1827308532 : SUCCESS in 11m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/539983fe17c04b18a3baf4b9317c4a12 : SUCCESS in 7m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c44d7dbaa214e1db52b2e36992ed80e : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/92b27611a05a4714bff418939b8305e5 : SUCCESS in 6m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d95b74dda3e14401b307766996d4c67f : SUCCESS in 7m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/81f6616053344042a2dfdc2df997a999 : SUCCESS in 12m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b0bd66ac5b445d7a01e0fca236442c9 : SUCCESS in 24m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e7f027e55bf4503aae11127c99a1bd3 : FAILURE in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f881966f8c3d4a6596b873fc3df33108 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/120307ef771140cbacbadadce15ad4af : FAILURE in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5153f933c8d84d2bbab7d9f1e0914453 : SUCCESS in 5m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8387fd2166df485ea291d2924724930f : SUCCESS in 6m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/034b332a76fb444696dbd50e18a7f315 : SUCCESS in 11m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/691341f7a76f4855a626e1256b47aab7 : SUCCESS in 53m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3a6f4b8c4c14e79a22003d5c064d837 : FAILURE in 18m 14s","accounts_in_message":[],"_revision_number":7},{"id":"1b468859dce300e436bb4f40346eba5520d33eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-23 20:53:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"105b43432576dfc8b33e67e0b11225ceefdc2b44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 21:21:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1012e9b1d19f4e85b0def98431e5ff27 : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":8},{"id":"ba021df0d05c0a340a491f5c0b77eaa4a1d7700c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 21:51:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5fa1fea1d5046d29a85730676436608 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a575d372ffd40329d5f73a6302cd061 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfbd9dcc328f411e86c489a253dd1a71 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93bf8c063a8428c9ec08dde09e0134a : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/922a9e60dc804bdd805d5e323197b6fa : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0107ca0fc3d8409a82053a3216b70eb2 : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/01fb4448d1364299b415385f9b90501d : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f97f9c673158439f9e3222e87df400eb : SUCCESS in 9m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02e9502f958a41d686fbcd88aa500209 : SUCCESS in 8m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ccacf3b218b646659762d759df5537ea : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6643b5e8bda45d5a10424e7ea6929f3 : SUCCESS in 4m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b4b3408430d24cd2883e999bc1f1e113 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a3bf5b56e5e342b5821059a41f873e0f : SUCCESS in 6m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d3a529f6460e4961a979e47b534cf75d : SUCCESS in 6m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/135f6abb238d4a448e7a27febac98c08 : SUCCESS in 12m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e4fc4a24774846b3baa1090aa3722773 : SUCCESS in 24m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/208af759e7c549ff9d03159f7699157e : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/222a43e94e55466eada62744730f8867 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e114f81f0f647e3a9301055938aa3d7 : SUCCESS in 17m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f85853b61a0a44248820568cea459c10 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59b3a30967024d92a8f5441f2254fef7 : SUCCESS in 6m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f61346510f37484c837dbc668d2def3a : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6ad43d392474421ad8ed88c9c6894a8 : SUCCESS in 53m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a936fe6089d7400ca453aad9a1a28c80 : FAILURE in 17m 55s","accounts_in_message":[],"_revision_number":8},{"id":"1e9836c674031cf47b706c57409eae97fe2225b1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-23 22:42:28.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"fff984a69b1ec77813a44fff59f0297c5d87806b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 22:57:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d06c06c7f51549258ebadc2adf8fc803 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":8},{"id":"9f17f24a3ca549538f1105ab06a8eea94c7bb17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 23:38:43.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc7c1983221b4c5f9d26e7aa67de4905 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1307496572504ba084fe89203ebbc0b2 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a01766fc8bc240269a75b2ee2c299cba : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cbb612cbaeb4e32a044b51b5b895520 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f8bf65b5d084c59aa7ac97962789371 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f53a43e0b24f4673a0578c33a5346506 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa548e6d63bc46f698ee9983fd93321b : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b4853fdb9a394b9ebf052d1b8fb8e4a3 : SUCCESS in 10m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/321e561ead2d4e3d9018653c1f12584f : SUCCESS in 8m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8344856a3a334ce5a9910b51cf4d6d75 : SUCCESS in 8m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a7b655fe4be84c689451a5b2bbf7cd33 : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c5f34cd6cdcc4ab68178f9dde8dfc83b : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d05ee8807344e9a9b39d2bb3db83ce2 : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b94b263fb0c04dfbbf4fffeceb285e56 : SUCCESS in 8m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/75332aee205c431f9eb043b5242f1fe8 : SUCCESS in 14m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ca0775d61da141808aedbbee028f9413 : SUCCESS in 21m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9851cc18ce354ccbb2218b8a0c3fa543 : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf416074f9c44f3fbe62e63853066810 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a132bbecdfac427daecd62db7a23aba6 : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5b68b6c58b9a4881b7d62491605da75b : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90706d04e77840acb0b1f51eb10f4e39 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/38bb56b7f2494c7abb7c5dff7111eb94 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4958e1390cc340aa81de09c93f584dfa : SUCCESS in 53m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30e6bec4b041411db55429f0b9a1846d : FAILURE in 21m 35s","accounts_in_message":[],"_revision_number":8},{"id":"1cfdada8f7cc744a07712fbdfc26af8eda76dcf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-24 03:21:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1564507d274000ae4cbc728915031e7fad85af68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 03:41:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f2d427d3c8f54733b58294eb5f0cac74 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":9},{"id":"ed7416eb3badd02562662db86ee4f550f32e2a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 04:14:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b36df220d94239b4dd84471de0ad38 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032f83ab99274b0e9618756305987d7e : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f3d6e07781441cda1d5db431dfb4b49 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdd681831d10475993f99fe7d119867d : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d0075d4f6cf44678f5ca9d0c450e8ed : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a38f7bb7d7849b1ae8d3eee7914fed9 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9bfddb7c8d9541fa842df9c6e4e6ba4b : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d77cf0c9731e490085767164ce3e35e2 : SUCCESS in 21m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce197f06d61447ab913d70385a118778 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cdb10ead9d30489f9f74c36f10268bcd : SUCCESS in 13m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/96338f2ba90e45fab750d7e8c4c0b241 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a29a85973f534a19ac288665f155bf83 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/366d7b3460854c98b6c0dbbbb6da2f1a : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1e783a6c33d146ab99d7da12998d27b3 : SUCCESS in 8m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ffab4ecf4f7e4e47ac0d529867934103 : SUCCESS in 14m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fe10f85adf454f7eb746b9e414909352 : SUCCESS in 28m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2f4615524f96476985dc5ec24680d3a8 : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bda70e76dc4c404585e9b6480ba26a17 : SUCCESS in 5m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2c3084ccaeb44a3db42574dd0280dc12 : SUCCESS in 17m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/08149e8c4c514c23a2b18c8c627405d2 : SUCCESS in 9m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ffb57e6c15bd49099af0152faf653907 : SUCCESS in 6m 47s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/baf043acf14648d4be6f8ff05a2bbae1 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c28d66d07f944314a117bf7e92ea148f : SUCCESS in 50m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/866d7cb1c6ae4b27aa43ac61417d57c6 : FAILURE in 18m 26s","accounts_in_message":[],"_revision_number":9},{"id":"50facd5752f48b3c82d17d0bbaea947345eedca2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 02:22:49.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"768b52b9aceb3b5e9b77f035f9ebca07b234e0fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 02:42:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/07c9bc07a02443a8ab8c2a988d42085a : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":9},{"id":"7485ad3ee51f9601612b02307f82939768adaca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 03:22:35.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4371cce6f05f470c8ccfeb5e2a0164f3 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ae7bd6557d475493410016937d542d : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b883b9eb1624aafb2e5df8f541ea817 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c10f3a38ce9f4444abdfd4a5ded03c75 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4ebf650d2d949729468bd477cbcc47e : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b17730d25bf4ef4af8bd29a6aee66db : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8891f3f344124223b6b81ab5776643c2 : SUCCESS in 2m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/423f62a4017b4d5082ff0979fee604d7 : SUCCESS in 25m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/431b0d14d68045599bb851fb3622ad43 : SUCCESS in 7m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8aee1543408a4a5f8746f1d3acaaba50 : SUCCESS in 9m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6babe4b7121447b8ae3330a3e34c9015 : SUCCESS in 4m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/232e148312dd470f87c52fcb7f0010f5 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea38d76069224cad9251e1a74cb38155 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3700a7690668430799d31df9993e5da9 : SUCCESS in 6m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/44f6628475e54b19b1200c213a694a4e : SUCCESS in 12m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6ab4267a71b840cd8dff519dbec57b3f : SUCCESS in 31m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6a06dbcacc654dc3ba6372300c0698b9 : SUCCESS in 11m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36c74f0a343a4168a5f3ab8731a20bff : SUCCESS in 2m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de5dccf8840640abb9f3e4a565eefc76 : SUCCESS in 17m 37s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bcc4d4d6239d4f67933cec4cb7eddffc : SUCCESS in 4m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68e9968c72f04bd7b682d168a3851e45 : SUCCESS in 5m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1bbcbbede9444a58e78a229ff28a1b0 : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88b4c3787b194152bc6215c5b2218957 : SUCCESS in 54m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51a7e50b566d4726a723ba86a7f42513 : FAILURE in 17m 24s","accounts_in_message":[],"_revision_number":9},{"id":"da40e0a2a8219207d8ac8c4e2ad555befd0c9f3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 17:06:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1023327465532f32a781f8667c0d35d526cf7f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 17:24:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ef56eea46ae44888857413c22b96cf3c : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":10},{"id":"6579e7a4a97ebf6cb1eed9dfe5a14014f727e6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:06:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b627a10b6e184f61812e33271139f318 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ece071010df4bdf8d86579595ea02c2 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c1179f34ebf4c4b9bc188cbe8360380 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62c8df70b294436faf6e937ac1a0a869 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ea8d0ec9de948daa313a9f7486e5319 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/37534b5d7c4040a09c2e8fdecceb3437 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe1dcd82ddce4f3383bf2da3242780b7 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/47f47848e905455ab1157e1f1b68cb95 : SUCCESS in 12m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eac02c09609443378db01df9c4794cda : SUCCESS in 7m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/98c57a489a3f41d09061dd805acf6df3 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a32b97051fd642489f61c81f2d35bc87 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e747966ca8f745ad80f7216a7ec98e51 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d8f84ef9361547c28dec1aefb98bf770 : SUCCESS in 8m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/58bd042f2a0e495589769c3a58da085e : SUCCESS in 6m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1ce0f49301ec41188653a6dadbe29eee : SUCCESS in 13m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d311204285924450a6a8fcf6f49c994f : SUCCESS in 19m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8f070f59fccc42ff87117339182b51d4 : SUCCESS in 11m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1bbbb9408956473197d07e50518ae997 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7cff319e3024bef83e9e52d9c91a264 : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/31d9a19bb9d84b808f0f3141a8e3df53 : SUCCESS in 3m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02b6eafbcc7c49aa8287ba81eb1e05de : SUCCESS in 5m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/67cbbbba81ed4123bf48de1fa77189cd : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc953154d5794d02896166da85ba0b93 : SUCCESS in 38m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0eb6111f5e214b8aaf44abab3b6bd4df : FAILURE in 58m 47s","accounts_in_message":[],"_revision_number":10},{"id":"9f568196c109d2f564e6c185a98f4db93758386b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 18:08:12.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"618f66db486b837d58c163c2831f0fa288e668fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 18:23:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/85d9e0ac1a124f2fa832ff22e1cf20ce : SUCCESS in 11m 24s","accounts_in_message":[],"_revision_number":10},{"id":"b0128daaef55531aac5145e2ee0a3aa4025dd65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 19:23:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd753798c5b4bac80a0a113bb208277 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2c06e3964840759decdd972b371c8d : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57b7aa02914148f5915681aa4273c37b : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/258a16e8c56e4d1fb3ef9cafaf04028b : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ac28c33242d4ff6951252c74cf4c03b : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/458b386799a34e20afa220a22b61456b : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cadbf90b5e824beaa1abd2dad3e49364 : SUCCESS in 2m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/30a0cedf82d042d0b1e729e6675ffa03 : SUCCESS in 8m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b5924cd1b874579b43148e388fa975e : SUCCESS in 6m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d553946fadbf4021a3af98e3b2a732ac : SUCCESS in 7m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0dac50231b1c48d78b8eb1b90a17ee0f : SUCCESS in 4m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d5fe1b3ecbfe494f88b92bdb5773acb3 : SUCCESS in 8m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29fa8b3b5007474f8c4d3821328506d5 : SUCCESS in 6m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/704ea3fc694c47189b96ba3df2b799b6 : SUCCESS in 6m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/522fce0722a84967a81a4a773cd399fb : SUCCESS in 12m 31s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fd35ad0be1ea4e3489a1381d1920ea50 : SUCCESS in 28m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c0e074fb045475d9d19692b2ecf26ba : SUCCESS in 11m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2554819357b84811ba2747b9b1293059 : SUCCESS in 2m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c06fa12116b04f24a9ea9432a11ca7ba : SUCCESS in 12m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e6832b0b306c45bab45af5c5a2fd5e2a : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a9c4b242a79947398cdb84bf541e3866 : SUCCESS in 5m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a2f97c17e5e846d090a377e290573603 : SUCCESS in 9m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6333ed51349e49888c3c0c1a110034d6 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3b40d40be594b409bcc07d6ad6afda0 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":10},{"id":"6ffb8caf809f0625bdfddba566d3efaa7fcff45f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 19:45:55.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nyep, was virtualenv","accounts_in_message":[],"_revision_number":10},{"id":"dffe87b2137a3b0e064a437baf70e4824444c640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 20:43:13.000000000","message":"Patch Set 10: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"851821bbbafce7ad1adf83786af7ae5bbcae7c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 21:18:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"371426041647e0f5db3bfca749f9353c61292378","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-25 21:23:50.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"b128deb625689c2d1cbec5ae0388d1364ca80339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 21:36:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/909d0fc086ba4c01aa3ac66fa6ebc483 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":11},{"id":"e51af3c606f84e6793804fdf4e0f6641787924da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 22:47:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f6584ba5c04d04a0db65b29fe5daea : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58ebe19e5fd4601bd2238aaec9bfc73 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/057536043a7d43938228af2ea472eee2 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a173eb6bbd534ca58e6c4b2ac2c72cd1 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/826c6f2f2ab948249253b1146891f0b1 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d2cc716a4334a95930004c71782587d : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6332e31729ba4a068c0dab691a986e6e : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2dc5752cb0674b98a21767c077aab2f4 : SUCCESS in 9m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48fa15b6db7548a186f8c5d01f969b20 : SUCCESS in 7m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d1979e38f51a4bc49bc966ee844851ef : SUCCESS in 7m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5aeb2948d4a6486695afabdf0f08d571 : SUCCESS in 7m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/df39ae3b61ae44819880d726e4d7b216 : SUCCESS in 6m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fb87f83f02c4b5894cea1d9d94e3fe8 : SUCCESS in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3834365a99be428787ead72e09f4ab87 : SUCCESS in 10m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d7807f2b0c0040f3bc3189c7226f44fd : SUCCESS in 13m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/939ea21e0baa4183a6e2e9167f9e2062 : SUCCESS in 20m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e6ee28909a4c4e608540344e7678b34d : SUCCESS in 9m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/02e5ca5488d54c22817cfe99aac159ef : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24ca9c0fa4c3459d80a672dc57b2345a : SUCCESS in 19m 05s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b37b8c01459849429edba0d7418ed6d3 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ceedbc2a71854e9d9a054a3df372167c : SUCCESS in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8dfd68262d29494b9f80db4db22298f7 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f9d0cd6aa4a4cf0925341afbaa0e8d2 : SUCCESS in 58m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ddb56886f45490c89d07f72f4ca201c : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":11},{"id":"6ab21f33bf1661e55219e8498e7aaba89dc577d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 22:47:21.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d6f07c78c62aa533374d944e30bd39a17135df3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 00:02:23.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2963206d6d47fa8a34f3ece48ab54b : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0e663ff4834756b6c9847dff4e2f69 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f41ffb4937da4d6b9674b4322222db73 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20a8d2c52d41458b9909d4584fd519c4 : SUCCESS in 4m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc2397fbd14443fc9faf18c178dc902b : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/229b87d20b3a4f78a0fdd7d96d222a9e : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/120fb2cda569471db2f9ca31edae1e15 : SUCCESS in 10m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7aa1e597411247788c4c3aac8d22bf1b : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e0b316d942b044849016df3ad30e86e2 : SUCCESS in 7m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3cfcac2fa67e4d7e8d639de57750080d : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45a074f960b641f69f583fa0631b8120 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7f9a9d7df7c24a6aa2d184211a4d22b6 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e3f155d2dc834dafb8c895967a5e6a12 : SUCCESS in 7m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3eea74ea5f4641a398dc4fe5e256dadc : SUCCESS in 12m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/753ae2eade0c4f85ade28a82001c542c : SUCCESS in 26m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f84b0a5c2b3f420384344fdf0bbbfef0 : SUCCESS in 9m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c7780f76d6ac43dd8355ab14a07a2c51 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e7866f4106f4f59a868c5d5fbc3386a : SUCCESS in 15m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e668afb3e2164946971b9e030c67cc7a : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/249afcafb09246da8715c49ec0996cd0 : SUCCESS in 5m 39s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cf8d1d94f2146d199c2e2c7658e1e35 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88bc4719df7b41868ea3f73fd15231f9 : SUCCESS in 54m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31fb0f3493ce49c9b98fecb5ff352706 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":11},{"id":"90193b992974a97198741a8646891ecf64574b64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 00:02:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"2725f49c2701cc870fcef240198b23a9c25f0134","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-26 00:03:36.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2d8151418446ddab72df3bce9005bb : SUCCESS in 49s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"25df9494b9a50269fbc3dca398d4c18d0bbaf5e3","revisions":{"dca5c5e62c39f08adceb3addd27268100c8a486f":{"kind":"REWORK","_number":1,"created":"2020-12-16 07:13:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/767280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/1"}}},"commit":{"parents":[{"commit":"42e95166ffe584a722697048c6a674be03caad02","subject":"Merge \"Exclude packaging 20.5-20.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42e95166ffe584a722697048c6a674be03caad02"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-16 07:13:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-16 07:13:18.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dca5c5e62c39f08adceb3addd27268100c8a486f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dca5c5e62c39f08adceb3addd27268100c8a486f"}]},"branch":"refs/heads/master"},"618bf1b389f0fddb112e8d35a552f52104c1081e":{"kind":"REWORK","_number":2,"created":"2020-12-17 06:24:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/767280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/2"}}},"commit":{"parents":[{"commit":"b12c8794e87d97652981f6843c5cdfc989128040","subject":"Remove Karbor deliverables from Reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12c8794e87d97652981f6843c5cdfc989128040"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 06:24:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 06:24:29.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/618bf1b389f0fddb112e8d35a552f52104c1081e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/618bf1b389f0fddb112e8d35a552f52104c1081e"}]},"branch":"refs/heads/master"},"26d57de530ff0fe5726bac6b91e4ad55c1deb98d":{"kind":"REWORK","_number":3,"created":"2020-12-18 06:26:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/767280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/3"}}},"commit":{"parents":[{"commit":"b12c8794e87d97652981f6843c5cdfc989128040","subject":"Remove Karbor deliverables from Reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12c8794e87d97652981f6843c5cdfc989128040"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-18 06:26:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-18 06:26:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d57de530ff0fe5726bac6b91e4ad55c1deb98d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/26d57de530ff0fe5726bac6b91e4ad55c1deb98d"}]},"branch":"refs/heads/master"},"afb8310c58435e29c68b013a12602c644da22486":{"kind":"REWORK","_number":4,"created":"2020-12-19 06:23:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/767280/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/4"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-19 06:23:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-19 06:23:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afb8310c58435e29c68b013a12602c644da22486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/afb8310c58435e29c68b013a12602c644da22486"}]},"branch":"refs/heads/master"},"fcef5963b45f4b9524677bc051510230c50857fe":{"kind":"REWORK","_number":5,"created":"2020-12-20 06:17:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/767280/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/5"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcef5963b45f4b9524677bc051510230c50857fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fcef5963b45f4b9524677bc051510230c50857fe"}]},"branch":"refs/heads/master"},"736a97a1c40eefeafd7f45ec1e60fc5515639d10":{"kind":"REWORK","_number":6,"created":"2020-12-20 21:23:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/6"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-20 21:23:35.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/736a97a1c40eefeafd7f45ec1e60fc5515639d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/736a97a1c40eefeafd7f45ec1e60fc5515639d10"}]},"branch":"refs/heads/master"},"4cad631458224880ec83111285154256463f7783":{"kind":"REWORK","_number":7,"created":"2020-12-20 21:28:25.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/7"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-20 21:28:18.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4cad631458224880ec83111285154256463f7783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4cad631458224880ec83111285154256463f7783"}]},"branch":"refs/heads/master"},"6ca0ff746da4ff08ad1bf288488f492aeaa02c83":{"kind":"REWORK","_number":8,"created":"2020-12-23 20:53:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/8"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-23 20:52:54.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ca0ff746da4ff08ad1bf288488f492aeaa02c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ca0ff746da4ff08ad1bf288488f492aeaa02c83"}]},"branch":"refs/heads/master"},"e0995e11e491d5d2adc8aa3e607a7f4ee5ab030b":{"kind":"REWORK","_number":9,"created":"2020-12-24 03:21:42.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/9"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-24 03:21:30.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0995e11e491d5d2adc8aa3e607a7f4ee5ab030b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0995e11e491d5d2adc8aa3e607a7f4ee5ab030b"}]},"branch":"refs/heads/master"},"3dd6ed34cb1aea6a84bb633d62a10601065cafba":{"kind":"REWORK","_number":10,"created":"2020-12-25 17:06:38.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/10"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-25 17:06:23.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dd6ed34cb1aea6a84bb633d62a10601065cafba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3dd6ed34cb1aea6a84bb633d62a10601065cafba"}]},"branch":"refs/heads/master"},"25df9494b9a50269fbc3dca398d4c18d0bbaf5e3":{"kind":"REWORK","_number":11,"created":"2020-12-25 21:18:30.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/80/767280/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/767280/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/767280/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/767280/11"}}},"commit":{"parents":[{"commit":"d8a0a38c313db5d4683987c41605704ee5fab409","subject":"Merge \"Remove the searchlight deliverables from req\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d8a0a38c313db5d4683987c41605704ee5fab409"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-20 06:17:43.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-25 21:18:20.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ib2d5f618aff891a7414efb382fb7eb1d0ff66c75\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25df9494b9a50269fbc3dca398d4c18d0bbaf5e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/25df9494b9a50269fbc3dca398d4c18d0bbaf5e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
