)]}'
{"id":"openstack%2Frequirements~767529","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I6a4679db864e516e82774f06b89c07b5c7047e5b","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I6a4679db864e516e82774f06b89c07b5c7047e5b","subject":"update constraint for tempest to new release 25.0.1","status":"MERGED","created":"2020-12-17 13:07:00.000000000","updated":"2020-12-18 02:38:16.000000000","submitted":"2020-12-18 02:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767529-new-release","meta_rev_id":"cb9a0da1cd1ab8ee718d0cb8f06004c8e6a6fd73","_number":767529,"virtual_id_number":767529,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 02:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-18 01:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-18 01:13:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 13:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 21:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 01:13:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c404a2b563824c8106198d2fac8eef2428c8409f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-17 13:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e7d8898923d35039f35260382dd26dbaf36d3ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:42:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2bcb926019aa436fa192bfd57d133e33 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3c84fa33fee4773c5e4a4040d4c634da03a6c87e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 21:12:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ba1a3eadc854340999d36fd7980d21f : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b6ee574b80341489e9756267e95d478 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f466e20580f447aa4fe465742292e23 : SUCCESS in 4m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3cce24384b0b458fb07b929c2347f6a3 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b5ec2337a56d4c06be980ca24c4c24ed : SUCCESS in 3m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efddd29ae7ea43d9930a9529b6f70f08 : SUCCESS in 7m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b00c44ebd5fa43968fa17bfedd9094fe : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/460b2858e9154150b0162a4a2d754ac5 : SUCCESS in 7m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3ced453801e34a07b3b61a9eb946d7f6 : SUCCESS in 5m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1ca8ceedb39c4263b1aa99f056b68da6 : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66b13c331fd84c3b9eb5721c3f23cef0 : SUCCESS in 6m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7954c8c4bd3b4665b69066c4b30cf14b : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5362f36b288b4aaab48335f02d0945d1 : SUCCESS in 13m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/78087aea8b564b1bb27d5c89a330184c : SUCCESS in 30m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/31f373d3c67846ff9254ece087caaec3 : SUCCESS in 12m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6927ae1adf7e49e1bc2fe1c9ecbd23d0 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/886cad1cf80c4582aac5dfdc2895d298 : SUCCESS in 17m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b9dc681ec0d6444aac4bd40c7312038b : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/640cc9305aca44e7bca3fb9ac921403c : SUCCESS in 5m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b102b1c6c8e941629e147daeb810ae8b : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4859a9a213a43eb868389316d07b4d0 : SUCCESS in 1h 01m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f099cc495e9c4187bfd31f1b1571dec0 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"eea10c737b89dcd66af15f6cdb50aead159adcd5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-18 01:13:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ae5e712af3f7c9f1d7bd465c0891fb8715aa332","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"347a5b90ea93d99176a6f0b77526f6efad120c11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:38:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e02b5510b894ccea659712d5568bc42 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d0e38145fb24e58b73488d3873bcf69 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c7dbb71b1cb4ea09f95c71684d02a37 : SUCCESS in 3m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1a8acaf6091416db6aa9c94e2041e2a : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/65fbd9b5a91d4654ab7aee9dda05ab0b : SUCCESS in 4m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6bbf7ad2a97e4baf9d246764aa7509a3 : SUCCESS in 7m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9eb78f6a1d934f4485039e326c448375 : SUCCESS in 8m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc97d52a07c041008ab7ccf48706d476 : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e4674c14a76e45c298e016b654e49e08 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/be7776d444d048629bf34aa5c8bbf677 : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5e3f35eff689475dbfe734e7fdc46031 : SUCCESS in 12m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/023c45e528704dfcb92775e4d2a02e83 : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/890550dbde5c48db9a0b33d3f70981f5 : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5fa747f32476491e9896185064455bee : SUCCESS in 26m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f371c1c928e34f7da9c0428635b30e89 : SUCCESS in 15m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8dfcfc1ab7114e75a98bc7aa771b3b96 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8ec16b8ce02e495eba022e52d248d55c : SUCCESS in 22m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a25b4c1f9dfe469f873c7b9852c6b372 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e248fde2cd204d21a72cd1f2011da784 : SUCCESS in 5m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/afd0550629af43ac9ff233f2985ba8ca : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fb8eb7915ca43738fb09f9bc7b2bf53 : SUCCESS in 59m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8882859ebfc407dade942cd155b6ab1 : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cb9a0da1cd1ab8ee718d0cb8f06004c8e6a6fd73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:38:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98af317524a9d01c5a4f866c01e8f6fc1c7462e2","revisions":{"98af317524a9d01c5a4f866c01e8f6fc1c7462e2":{"kind":"REWORK","_number":1,"created":"2020-12-17 13:07:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/767529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/767529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/767529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/767529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/767529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/767529/1"}}},"commit":{"parents":[{"commit":"36710831f3154ec0844fb2a38317815435da46bc","subject":"Add neutron to stable upper-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36710831f3154ec0844fb2a38317815435da46bc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 13:06:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 13:06:58.000000000","tz":0},"subject":"update constraint for tempest to new release 25.0.1","message":"update constraint for tempest to new release 25.0.1\n\nmeta: version: 25.0.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Commit: Ghanshyam Mann \u003cgmann@ghanshyammann.com\u003e\nmeta: release:Change-Id: Ibdb2d8cde5235e1f1479b7a09f579eae04276bea\nmeta: release:Code-Review+1: Doug Hellmann \u003cdhellmann@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Masayuki Igawa \u003cmasayuki@igawa.io\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I6a4679db864e516e82774f06b89c07b5c7047e5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98af317524a9d01c5a4f866c01e8f6fc1c7462e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98af317524a9d01c5a4f866c01e8f6fc1c7462e2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
