)]}'
{"id":"openstack%2Frequirements~767530","triplet_id":"openstack%2Frequirements~master~Ib13266420fffaec8ff62f82098a61bb7732ea8b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib13266420fffaec8ff62f82098a61bb7732ea8b9","subject":"update constraint for metalsmith to new release 1.4.0","status":"MERGED","created":"2020-12-17 13:09:38.000000000","updated":"2020-12-18 09:05:58.000000000","submitted":"2020-12-18 07:46:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767530-new-release","meta_rev_id":"8c2b2266f19d8f1b1edc2cc66ec93338deb3274b","_number":767530,"virtual_id_number":767530,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 07:46:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-18 01:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2020-12-18 09:05:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-18 01:13:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2020-12-18 09:05:58.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 13:46:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-17 21:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-18 01:13:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-18 09:05:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"0c6bf2c42f18f9f19bbc7d8f3f9b6ffd692e7d5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-17 13:09:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b59309a6d7139a76690e200874210e1b9ba4753","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 13:46:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/823325d802bc4b2595da60f6cec1c0f9 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9eb6e4acfde6b79c094526fcdb07bf40c1d0064c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 21:30:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427809e73c7e450680f6133992c587df : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2740b061ceac4397a5d310392717f607 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/369911938f9046eb995398452c484d6b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab72b11127984a9c8affefabca3d58b4 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89f3ef521fe84706ad1f4116cd674e21 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90099bc4a7534f56bce103da229655ea : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82c78f8ac3be4b3fa6266a3eb67414a0 : SUCCESS in 3m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/484457a22f9e445eb58640a9a435a99f : SUCCESS in 10m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1b5bfe7c5d2640d59a168b3c76dd7e19 : SUCCESS in 8m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0e1bf04685ad4a1c9b858e732ec03fc8 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/873ebb7d3b5d42308051ae772dd08525 : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5736cf9652764e2aae47dad39a6c5f25 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee0ea83bc9944ff989b0e5681fc13ef8 : SUCCESS in 6m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1bbf5b10a9f44e399b4860b5f3f102ba : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e3be63ef69e4306a671ca4272ad50c5 : SUCCESS in 13m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a3d3b189e0dc49a686527fa446b43754 : SUCCESS in 26m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/048d2b72c67f434e851c991f1160a54b : SUCCESS in 11m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1da426e2b8f14a8b82727d3df7b0e564 : SUCCESS in 3m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a593498e03194c1db05a09bd113748e3 : SUCCESS in 19m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f77267232ba486e8a9e0479d1ee1ce9 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be9c58969aa54b3ba43cc1691ccadf3f : SUCCESS in 5m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/10fc547e7d9c4891a190d44b906e975c : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03075ec9a8a94eb5a0629661c67660cf : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0db0d3a02655456da5defd8ffc810abb : SUCCESS in 1h 22m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cedaeda63a7a6e0cf751954426211937c39c085f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-18 01:13:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c1f26e89c9983e12f54302bbff6e160c777bbce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:13:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34f0dbbd5cd042d8ec7aa1ea3dc5e4101d1e7c71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 02:42:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bffabfd1cdd47649ae1280452f242e4 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25781710a7c8435a9163f01cb6645e31 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1031bd19063f466e99285a1a5e3e3f77 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7281eb0f51b645f1b123f972b9c2903a : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e2e27269e5f4cd8abe72c559f3fdafc : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/834d93d758ad496986707f5ad1ff197a : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87bc6fb26e6046c9913bc0f9da0dfc00 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/658a079eef0e447aa9f8bbd18abdd3aa : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1b2ded02729e493899aae22811e690d8 : SUCCESS in 10m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0eea3213c8764ec79030c98155d606f0 : SUCCESS in 4m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/18b15f9b80774249bbeb0bef8fe60fbd : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/728f4a4d853d49fb9df3315f4c0a7f0c : SUCCESS in 6m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e84b3c09d9f24e41ac626d7a3cb9240b : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8a335e2c582240e09910fcde2354be56 : SUCCESS in 13m 51s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4d7175fea7bf4a1eaae53e7f08a92fc2 : SUCCESS in 29m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/898319d2d02b454cafed2cd9fb01e035 : SUCCESS in 11m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f827aaa2d87d4f1f92c51b422a90cdf9 : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9e0f74dd2634429cac75fb01daf8447f : SUCCESS in 17m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/beab5928636f408288901ad6689acfb1 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/845c268b2aae4730a23de661b041e344 : SUCCESS in 5m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2136bf4f1853406b9c89bb8111ba70d2 : SUCCESS in 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a005943368134da5aa29a8e118ca9935 : SUCCESS in 59m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13151f023bcc4c67abd7f3aad44b8062 : FAILURE in 1h 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"aae965237cd8b4f2581edebd105eb0d2dd6e4840","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-18 02:50:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50a09e5cf4f21ca342a2fa1077512b3a61d87b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:12:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bf79d760bc054045aa1577fcb8f1b091 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6f30a4632e94d00a8cc4bd42ddc0ec497bcd62cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 04:19:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729a53c1d49d4717a4d36712d7622ea2 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760132a8b9624c608428e959349df126 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fa357cc8d0c4180848ac7c3e7a97a9d : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e892fe422c4767b3e6201363496f1d : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0c942af6d2c46e380544044bd43f060 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/680c946c6a09445aad0e4c3b389fb180 : SUCCESS in 3m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e3e1bbfc346d46a4b682d50f0589a918 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/558708a37ecf4108bfd21f14da6e409b : SUCCESS in 8m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c3c952063774193b9a58ebc947bf486 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/486b531751584af4931104a3b1ad4c26 : SUCCESS in 8m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8e83e90c46094613881d17d836503c6d : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0cdbc1ab149b408597fc3eaccdffdd3f : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e4cc0017c3854fa8bb38559aaab96555 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e73906e1513c4aa5bb24eb119de44e20 : SUCCESS in 7m 35s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15be8a7e1e4c4753993c66827cb7b276 : SUCCESS in 11m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b01c9a6ca4394d4bbc5b1ba8c6038971 : SUCCESS in 24m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/301e314a2dbd4298975569281108f9b3 : SUCCESS in 11m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c8cedcad2cb4ea38226e963fe89d75a : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bdb763fde91a4369b4cc4927260cfbd0 : SUCCESS in 16m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/40df0928ebed4c3e825b78ae86a7e550 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b087849edf994ac5bbe73abeb35b541a : SUCCESS in 6m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/272260b279324b90ae54937f90882f25 : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87ffad4eae834e4296c0a514eee12441 : SUCCESS in 57m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d19306d773794c05900dbf7b6c5604ca : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fad82c2b05323fda9804f34ca80009d2f973fce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 04:19:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4508dca50d40691769909186bb402a3fab141b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 07:46:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de405fe58954360a5a5ec948b0ad8e1 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39a4ca634ab4405caed9d77b4f2ef6a1 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5020287ef5b24e7da701dda13cffffb8 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72664039c7704508b45ea55f8a22effc : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d433f22024964b6692f0d4a96d6678eb : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/161c7329425a45f4ad59588d6f5ed55d : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b63731a9f8214378ae60ddff64a29617 : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dac65fabbd154701b64f392fc94989d4 : SUCCESS in 8m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4985acf401f34c05ac1bacabf31a633f : SUCCESS in 13m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9b85231694ad420e8760c0980c60ddbf : SUCCESS in 5m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92bfc0f892444aa283aff55260972e54 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b5a21a150764a6f9db9cd4d60dd1ae9 : SUCCESS in 12m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c014a0e9090f4a8c96ef775f6947c270 : SUCCESS in 7m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/96abee9e6e974670aa28a02090f616c4 : SUCCESS in 14m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/41dfdaa70b2f44f586205477fd942496 : SUCCESS in 26m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/912b22e875b1429faccdcdeafdf616df : SUCCESS in 11m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c42cb142042c4cdcbb94c4efa70756ba : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f7a14eb479049f2a888986c2d6fc93b : SUCCESS in 17m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ade45b5e17c42d5ab0d098b14fefa8f : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/00c1a9c5e26445d78c6d908038163c2a : SUCCESS in 6m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/265c2c47794048bfb576f2f92e66507a : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1aa6bf70788f432ab11395d89f0a6c6b : SUCCESS in 1h 03m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88d026273b104d9c9bc7f4884929cb95 : SUCCESS in 1h 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"01707f5c651b531ac70e99a84f579e9b2abce180","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 07:46:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7212cb50aac270be1cf1d042a1252053275f60c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 07:49:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecda8fa54b39421d8ab48def4fd7c012 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8c2b2266f19d8f1b1edc2cc66ec93338deb3274b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-12-18 09:05:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"230be2704b79243eb25a3c58ee86e6028f73fcdf","revisions":{"230be2704b79243eb25a3c58ee86e6028f73fcdf":{"kind":"REWORK","_number":1,"created":"2020-12-17 13:09:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/30/767530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/767530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/767530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/767530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/767530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/767530/1"}}},"commit":{"parents":[{"commit":"b12c8794e87d97652981f6843c5cdfc989128040","subject":"Remove Karbor deliverables from Reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12c8794e87d97652981f6843c5cdfc989128040"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 13:09:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-17 13:09:36.000000000","tz":0},"subject":"update constraint for metalsmith to new release 1.4.0","message":"update constraint for metalsmith to new release 1.4.0\n\nmeta: version: 1.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Steve Baker \u003csbaker@redhat.com\u003e\nmeta: release:Commit: Steve Baker \u003csbaker@redhat.com\u003e\nmeta: release:Change-Id: I844f794aa928a96b32290aa4e9f6e4699bfda2bb\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib13266420fffaec8ff62f82098a61bb7732ea8b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/230be2704b79243eb25a3c58ee86e6028f73fcdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/230be2704b79243eb25a3c58ee86e6028f73fcdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
