)]}'
{"id":"openstack%2Frequirements~768148","triplet_id":"openstack%2Frequirements~master~I8d18c294ae2262d94dbc7a483bad4518bdb97807","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I8d18c294ae2262d94dbc7a483bad4518bdb97807","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2020-12-22 06:17:43.000000000","updated":"2020-12-22 16:16:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58fcdb0eea26744d9b4827488439ef05b591b7e2","_number":768148,"virtual_id_number":768148,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-22 07:38:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 06:31:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-22 07:38:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5df78e56fb03aafc699cc8b62e68071ec887ef27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-22 06:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37311a6130f2318d498352c9f145eba9eacb046b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 06:31:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/22d697f1f0da423a89726c0fb158fb59 : FAILURE in 9m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9bfce3e4b0e7e626a8153a2ff0467e2033f579a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 07:38:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bd0324425d4877beb85dc5be2a2037 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3bd41fcccb34024a8e3e503e6664df8 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29250126c9974ab0a4dda3a819a5a913 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb909afdaa04110ac054b7a06614ffb : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56d131ea4ea4243989c75959bfa726b : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7510ec9a0624fba88bbc14689790058 : SUCCESS in 3m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4798c2f7f2d842e19bc5feb043f1863d : SUCCESS in 3m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6eb6e7216702441caedd3ca7269d9d6e : FAILURE in 10m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7fc7b4f320eb40728248fc74a9b4d317 : FAILURE in 3m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/476f185d9698485b9f48c47c90acb3fe : SUCCESS in 15m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/964c59cbae3c472ebfcc503eaec0d9d2 : SUCCESS in 4m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/208fb252fc484645b70d5d9083fefc69 : SUCCESS in 16m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3c781df8ee149ea84b3d803e72873a0 : SUCCESS in 7m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2686e89d97574c418e5f72844f0bdd3a : SUCCESS in 7m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bde889560c48421c9ebc46bed5137abb : FAILURE in 21m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c403407b4e26465d98d6f075dba139db : SUCCESS in 26m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3a1dc88b09249989d750e54fe82a8a6 : FAILURE in 12m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7abc3dccba1e4bd9a856b303a7d756fb : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0ff58f4f807457a9bceef067b4f4bf5 : FAILURE in 17m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/34591c64ce554dcf92a7adf064133fca : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5393800415c14a099fbf59e589ecbb86 : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e4588c2447dc4a25a32f0db503533e8a : FAILURE in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18aab424a574498e99c0f8fb4e20cdd1 : SUCCESS in 52m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8ba8bd3af8f48779534e74f21da6ced : FAILURE in 17m 41s","accounts_in_message":[],"_revision_number":1},{"id":"58fcdb0eea26744d9b4827488439ef05b591b7e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-22 16:16:36.000000000","message":"Abandoned\n\nnoclob","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d08806f45845b303b7999b9a6f3bccc44992cb7","revisions":{"4d08806f45845b303b7999b9a6f3bccc44992cb7":{"kind":"REWORK","_number":1,"created":"2020-12-22 06:17:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/768148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/768148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/768148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/768148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/768148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/768148/1"}}},"commit":{"parents":[{"commit":"4bb5643ad24d1ed49e5095ad1b8aefb168b12b12","subject":"update constraint for tempest to new release 26.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4bb5643ad24d1ed49e5095ad1b8aefb168b12b12"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-22 06:17:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-22 06:17:41.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I8d18c294ae2262d94dbc7a483bad4518bdb97807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d08806f45845b303b7999b9a6f3bccc44992cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d08806f45845b303b7999b9a6f3bccc44992cb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
