)]}'
{"id":"openstack%2Frequirements~768673","triplet_id":"openstack%2Frequirements~master~I04adb80d2c12394110d1c85c7312d2feb84e9f63","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I04adb80d2c12394110d1c85c7312d2feb84e9f63","subject":"Updated from generate-constraints","status":"MERGED","created":"2020-12-29 06:26:32.000000000","updated":"2020-12-29 08:59:54.000000000","submitted":"2020-12-29 08:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768673-openstack/requirements/constraints","meta_rev_id":"00e475dc916fa961e32e31957d2179453dfb40c6","_number":768673,"virtual_id_number":768673,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-29 08:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2020-12-29 06:48:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2020-12-29 06:48:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 06:38:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-29 06:41:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2020-12-29 07:51:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4bba514a2bdc61ef3d80cc8562f36e7920130888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-12-29 06:26:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"346ed6d79e80df68027398598c9bf8826aa8dd8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 06:38:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a8083d9b6a360c486ac336e5ccf9d2b80a4453b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-29 06:41:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"481ff58eb59efb6104ea3266dbf9bd1504f32799","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 06:42:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"dc81279fb4565b37e3c831013aba7fd7da5a22d4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2020-12-29 06:48:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fe110a468731441f494924fef59bf811e823dd9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 07:51:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd94cc92120940229b62bf491736940c : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfede77cec7461aa6fd23031bfcd719 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c9ec5210a08444187920816ff4e4875 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff23fbea5fe14369b49611c712a2c932 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c70372a915354e949fedcdf505d8830b : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9aec36738d504218b4a30001ef0e17d3 : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f3b12a9f37f4fff8457cd2dba798ad3 : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7b3e4824778a4d929d6a99ae481aef11 : SUCCESS in 9m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/952f9974a6ba4074a779e770a5dc76ee : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/feabde7773d7481eb70bc1205e9b3d3a : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9fc0f222825c47b390d0ad00653ccd98 : SUCCESS in 5m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9503e7d9c54145249a7210e096db2963 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/72b16949b8444172b6f7d1821ef1c439 : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b013ab6599d544138ca32bd557157bbf : SUCCESS in 9m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6ea7ced4545d436c9c50ef23d7daaaac : SUCCESS in 12m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/169959a5da81478582da132acd867ca8 : SUCCESS in 30m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f03d457c285b404f94ed4aba620d4cae : SUCCESS in 11m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55dbc03935de4b79827a4d54bd455215 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99bf9936d8bd46de9dad2cb2f4b02ab8 : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9c4fe0a1c414b04bb77ae4258ba0974 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c01f74fc05ca430489f2e82b76fdb089 : SUCCESS in 5m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d251b8ac69164534b48235a70021b0f5 : SUCCESS in 9m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/635e7fc1cdd941efbeeff443ea1667ea : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4eeeb1b710845708f26d0180e34f8a3 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":2},{"id":"c46b6e07e92c2d8985fd293d1ea4c50ba324eea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 07:51:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9978e0ccba068f457faff0ba73f9d4835b582863","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 08:58:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc70c78c1c4f4c58935addbc9ab33f96 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0d613605fd472d9b7999ceb9d95de8 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff0e0745523045c7b14e08ee2fdb6009 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ece01b4d04f34825b5649c4e2f9da1d8 : SUCCESS in 4m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cf82fc8dd7b74f30bd6e95eca50a99b6 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4052a73fab7b4ea78dd02ffd70e88188 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2aeadb0e21f47deac81b2b8ccd9b0fb : SUCCESS in 7m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c792ae4606c44dbad9f7d7e31f819ca : SUCCESS in 8m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aa6d0bc6526d4ca79d2f7dfc7534b889 : SUCCESS in 8m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a30586b7bd2b47319af4f165ec1e1017 : SUCCESS in 7m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f0cb1c85331748fcb4d5b215d59d7d5f : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0315897f5b454342803ae4415ce00708 : SUCCESS in 6m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23e0f7abdd7e46618e194c898f41f113 : SUCCESS in 6m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/916eba421ab54e9a9f7bcec0388871e8 : SUCCESS in 14m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/345a85e6ded44f7fb1ee5072b57b9df7 : SUCCESS in 28m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cfb6dcdaa9b1439e9ad50fcf41720e4e : SUCCESS in 12m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5852bd65ee0e437fae0f1c6a9973100c : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea2b99dd586a42588cfc727827b6578d : SUCCESS in 16m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8e7e2a29b3c9467e884ead33ce733def : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/001864619d27419fb025314b57567cee : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/763b9fe5c00c48b190127a2cde56bc3e : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7415ae4a08344aceb1b7e00fbb956d28 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87be78d0a448422992b3c15228414d74 : SUCCESS in 1h 06m 09s","accounts_in_message":[],"_revision_number":2},{"id":"756a96b045a9e512896beaf9543d1fa64285b0f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 08:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"00e475dc916fa961e32e31957d2179453dfb40c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 08:59:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea41cad5167342ee944c3ff002db762b : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ddf78ea56a44fa9380afeafc55e77e7325ab864","revisions":{"c64941f0538f9e9f76985f82f55391ae53160c8e":{"kind":"REWORK","_number":1,"created":"2020-12-29 06:26:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/768673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/768673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/768673/1"}}},"commit":{"parents":[{"commit":"5c23c4fcf81f79da2af4c84aa15562dd3e623fc8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c23c4fcf81f79da2af4c84aa15562dd3e623fc8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-29 06:26:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-29 06:26:29.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I04adb80d2c12394110d1c85c7312d2feb84e9f63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c64941f0538f9e9f76985f82f55391ae53160c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c64941f0538f9e9f76985f82f55391ae53160c8e"}]},"branch":"refs/heads/master"},"1ddf78ea56a44fa9380afeafc55e77e7325ab864":{"kind":"REWORK","_number":2,"created":"2020-12-29 06:41:14.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/73/768673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/768673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/768673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/768673/2"}}},"commit":{"parents":[{"commit":"5c23c4fcf81f79da2af4c84aa15562dd3e623fc8","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c23c4fcf81f79da2af4c84aa15562dd3e623fc8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-12-29 06:26:29.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2020-12-29 06:40:54.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I04adb80d2c12394110d1c85c7312d2feb84e9f63\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ddf78ea56a44fa9380afeafc55e77e7325ab864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ddf78ea56a44fa9380afeafc55e77e7325ab864"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
