)]}'
{"id":"openstack%2Frequirements~769121","triplet_id":"openstack%2Frequirements~master~I5aa1b194f980e5fcdfdd5052d2df334accc025d2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5aa1b194f980e5fcdfdd5052d2df334accc025d2","subject":"update constraint for os-win to new release 5.4.0","status":"MERGED","created":"2021-01-04 10:23:11.000000000","updated":"2021-01-04 15:37:44.000000000","submitted":"2021-01-04 15:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769121-new-release","meta_rev_id":"a21f9ecb47ec7d6ab9d0dcd5ec69c2b9dca7f04f","_number":769121,"virtual_id_number":769121,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 15:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-04 12:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-04 15:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-04 13:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-01-04 13:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-04 10:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-04 11:45:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-04 12:58:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-04 13:25:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-04 15:10:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"98dbe929d089ae388175d6333f8d0466780a0ad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-04 10:23:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ce5f1b73a430a89b1f2fee955553fac7facf767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 10:49:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3c35c89c2bb44017ac41069198069fda : SUCCESS in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6d77b544de0609d07404f421050eb08886afcb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 11:45:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae264e793fc414297ee383d493dddfb : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f26598892f40178c2c8306e7372db1 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18786f73f2fe43939c3f8ace4d05be96 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/865677e959dc43a3a9eca877befe7272 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/124a6a845d1e400294530993f67b2157 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a209daa5d9d24b0591de3611b1677f66 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/70290430ddbc4e80bfc9adf6c6d86464 : SUCCESS in 3m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0e4d14412834b79994bb4f073e840a8 : SUCCESS in 8m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7410b456078145f6ba9576a0f25b8b04 : SUCCESS in 7m 37s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0ecf7376f28743aabec11a247419184b : SUCCESS in 8m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ed0d2577415143c7826596f7d4675cf7 : SUCCESS in 4m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d8d16f72a1864820a23b679fb7f47cb6 : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7a0f1d8879a043d3b57a982b8084feb4 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6e75ec49d2b34dbf979df2d72e571b83 : SUCCESS in 6m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e7b5390ac6c4901afc323619b0bd78d : SUCCESS in 12m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f624f07f86e04a9d869357227f6ee2ee : SUCCESS in 30m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/54b8d77aa525447992827284907d765a : SUCCESS in 12m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86169f3bea014747beb42b463edaf232 : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57210a4e16e54858888f38a20f90d030 : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2a3fd7709a28427c99adcd0c7576f1ed : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be7ca219c65a4afb98b7d6658dbf2979 : SUCCESS in 6m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9a017a0e1f6944caa3622272fd018966 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7793c377b2b443bd85eccc3bcfd34887 : SUCCESS in 54m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d7d3d5730144e31bba76d07639141ff : SUCCESS in 1h 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9c2f5002248d550f17161add26dc2e4eed0f735c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-04 12:58:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b3b3c651b9016e23b35bcd1cb70d909ff2da28a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-04 13:25:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b83975a2c3380c990e439135802a9f598cdf6903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 13:25:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e9252d85247c973e1a111a4a01d2d67f183bfd9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-04 15:10:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1ad3f6a526c981cc93d1006f84a43e6353b2d46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:35:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41aecfebf564248a6775bec0f1c5ad1 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3de8d17bec6047cf88874a373b316366 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01208ee690e4fe5b98dfd21e64ff900 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57d5905e7e4f478ba593739fa8ce5517 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/daa5dbbf663b4ac3a727f00cf6d7b1e6 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad8e869380794e39b9438f2a65699973 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/488d8e10c9d14aae8842e60e4a598abb : SUCCESS in 9m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b635880c8b59490181a8deaf9aead111 : SUCCESS in 8m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/be27e2cbfe7c40a8b6f2b7996b1713fb : SUCCESS in 9m 23s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3aeadc0d75214c7eadafeca4910b1670 : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1d3231b2272a43b0815f9d9b0caf1723 : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/820f4f4223264618a63702a3ff59b860 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/54c51712e1aa45dda7fbf0107e8fa88c : SUCCESS in 7m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d1f55fff4f164f7b9aed62676d7da5fd : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6d471b39d29a4523ae8e4f191074b923 : SUCCESS in 36m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd741270a5d24e5da5c981deb83eb18f : SUCCESS in 13m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/266d112cf0944d08a04692508b349879 : SUCCESS in 3m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d52d99c442764de8846d248efa8c612d : SUCCESS in 17m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/77b1caee8ac3411bb4af266ae0b5a260 : SUCCESS in 5m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abdb9d2661474672b7711be3b4fc5432 : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8c17c54f4f8410c889af2b07262aa28 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1ff4f601db54e2cba13dc6517ec37c4 : SUCCESS in 58m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57a2bd9a758c48f1bdc293564a434f21 : SUCCESS in 1h 27m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e30ead0db3e91d91862fa109ce90a1c8c183c1b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:36:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a21f9ecb47ec7d6ab9d0dcd5ec69c2b9dca7f04f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 15:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72fa0702b1af4dba9e47d58d1b3e8e5f : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33ef3a76c72807ef81ee8e5dc7e10e7389ff3f62","revisions":{"33ef3a76c72807ef81ee8e5dc7e10e7389ff3f62":{"kind":"REWORK","_number":1,"created":"2021-01-04 10:23:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/21/769121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/21/769121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/769121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/769121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/21/769121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/21/769121/1"}}},"commit":{"parents":[{"commit":"8d035c52a99666ee7a56d79e139a8bc00da09308","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8d035c52a99666ee7a56d79e139a8bc00da09308"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-04 10:23:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-04 10:23:09.000000000","tz":0},"subject":"update constraint for os-win to new release 5.4.0","message":"update constraint for os-win to new release 5.4.0\n\nmeta: version: 5.4.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Commit: Lucian Petrut \u003clpetrut@cloudbasesolutions.com\u003e\nmeta: release:Change-Id: Ifc0686d216f94569fbf3eb61ad41c46e3c9a31cf\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5aa1b194f980e5fcdfdd5052d2df334accc025d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33ef3a76c72807ef81ee8e5dc7e10e7389ff3f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33ef3a76c72807ef81ee8e5dc7e10e7389ff3f62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
