)]}'
{"id":"openstack%2Frequirements~769455","triplet_id":"openstack%2Frequirements~master~I88deb79c47ae9ec91b038539780a38e9262a4185","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I88deb79c47ae9ec91b038539780a38e9262a4185","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-06 06:29:09.000000000","updated":"2021-01-06 10:07:14.000000000","submitted":"2021-01-06 10:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769455-openstack/requirements/constraints/noclob","meta_rev_id":"0a950e25cece1b4c78167d9c5bb65b583d5b497d","_number":769455,"virtual_id_number":769455,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-06 10:05:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-06 06:34:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-06 06:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 06:34:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-06 07:02:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-06 08:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71617e09e860e08d493eba8d554bc7ed7bd2e905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-06 06:29:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"784b6b761b06cb97b4c48d9189e5d31e2765dc06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-06 06:34:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3859a83ec60fe3560c4b34e7ddd687836600f102","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-06 06:34:25.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"3a88df8933a67d9f4c33a54fd6a582bc91f57584","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-06 06:34:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c62539779c30460912edef257c8563ccd42106b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 07:02:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/516b1bed7b15473fa023b001edcc7178 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d713701f9178af960b0399650d87f473c9d5d9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 08:23:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944e7a28d8804745830e5d1a0048b9e3 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256393fe60ec42c9b88de05ef3507fb3 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e370b54f20dc46b1a3e4e532ffdd4350 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c880659df18f4ca3aec32aad9a0e5a84 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad0c26377c2d4c95a234f848a2e3d4fa : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/19644e5f4dbb4057ac49832ab24affda : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eccabcc87ecf409a826a1d8d36fb1528 : SUCCESS in 5m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ee66fb2c52e4d358744f28f3f8005dd : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38d7dfa5a258465dafa5c259accc26c2 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fc1926369f464a8fadde7e1d2b9bde4a : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b213f5a6116a4033915f6a9097ede63f : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/44a94587f72e400580f9a52fb39cdd54 : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae4b3a2fae5648ffbb019455ef1f7220 : SUCCESS in 13m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/787ffb07c71c45f39d0bb6a32f7835c7 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9ceb0a1a1e214ae4819acf12798df29a : SUCCESS in 15m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3309b709b36647ecbafaa6b3c69e0a81 : SUCCESS in 28m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/81b9be33cd024f868bf86f84a29028ef : SUCCESS in 10m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b58bb29f293c4d90bb27c1be93acc950 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1e22af5902e444cda7de12f733457aa8 : SUCCESS in 14m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/44bd829e5b90418d8f6739c5a74c973f : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76bbe49fc39a46089412b3f2fbd79be1 : SUCCESS in 5m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/001fc9b492e142f19507715e02ebdf04 : SUCCESS in 9m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a905690c71a9456b83fcd70792b4dbd3 : SUCCESS in 59m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/090f4a5ed6944e3a80ea285cffda97e4 : SUCCESS in 1h 17m 20s","accounts_in_message":[],"_revision_number":2},{"id":"423856037f47f5ba6947da62566186a0bbc949b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 08:23:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3436e7ca4e00b03efa5178c5f82b125c122a26fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:05:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75ee29dd8f5c4e0db7f49a27057f9271 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74d36b2ddd1e435d93243868a76e6d50 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/223573555d164bfea14f41c8a4764713 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9b58e914e564ffbb17ed5c69acad267 : SUCCESS in 4m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c7d3ea801d940839663faccf7fc945c : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5473292716da4d7ca654563154ef3636 : SUCCESS in 3m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71a68e334ea349a084b50c67df6f613d : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4105723ef6a943939812e40246a00039 : SUCCESS in 7m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6917cac95db34acaaa310d858427b909 : SUCCESS in 9m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/12d2fcf0662e4dabb60226964ca64937 : SUCCESS in 5m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cea1b0016fd94fdc9012aff4bb0c98af : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9eeff3e1d9684ad4bdbb8a5dfbb26101 : SUCCESS in 6m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5345e58c43ab4d518fd3780ed0ee7484 : SUCCESS in 7m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/82ea8fc4f55847c999358906a7530172 : SUCCESS in 13m 05s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c4ef6d5db2b048ea913a4f2ea83cf75f : SUCCESS in 33m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f4fb4b13bf1e4d9e88d05849dd9e32e7 : SUCCESS in 15m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2260e3aa1bfe4366be87054372c742dc : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc11312804494de08409e22e56a7adca : SUCCESS in 19m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/80d6abc1212a49babbeff92dd24c7f5b : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a73422ac0e94f978fea24539e87e8d9 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5164e5f2ac354b84bb5c3b5a49ab3734 : SUCCESS in 10m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9323a103e574b559ab453dd0b305c9b : SUCCESS in 54m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35b1520447f447dbb2665e5bc530fce0 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":2},{"id":"2ecc3b27367fc1558703d87a102b148540290937","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0a950e25cece1b4c78167d9c5bb65b583d5b497d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 10:07:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09641c2a1f4d4ebf984bcffe0b878dae : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff08aee41973835d07a43c1b57363ef2e5e1f7ce","revisions":{"bfe65b31c9f440efda46c23c9160933e5beb7130":{"kind":"REWORK","_number":1,"created":"2021-01-06 06:29:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/55/769455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/769455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/769455/1"}}},"commit":{"parents":[{"commit":"4b9c97a1159fc80f3758571aacb855650bc6ffc6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b9c97a1159fc80f3758571aacb855650bc6ffc6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-06 06:29:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-06 06:29:06.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I88deb79c47ae9ec91b038539780a38e9262a4185\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfe65b31c9f440efda46c23c9160933e5beb7130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bfe65b31c9f440efda46c23c9160933e5beb7130"}]},"branch":"refs/heads/master"},"ff08aee41973835d07a43c1b57363ef2e5e1f7ce":{"kind":"REWORK","_number":2,"created":"2021-01-06 06:34:24.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/55/769455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/55/769455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/55/769455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/55/769455/2"}}},"commit":{"parents":[{"commit":"4b9c97a1159fc80f3758571aacb855650bc6ffc6","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b9c97a1159fc80f3758571aacb855650bc6ffc6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-06 06:29:06.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-06 06:34:15.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I88deb79c47ae9ec91b038539780a38e9262a4185\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08aee41973835d07a43c1b57363ef2e5e1f7ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08aee41973835d07a43c1b57363ef2e5e1f7ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
