)]}'
{"id":"openstack%2Frequirements~769571","triplet_id":"openstack%2Frequirements~master~Ie80e45571efafe73a553ddd89c5be90fe97bea4d","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"Ie80e45571efafe73a553ddd89c5be90fe97bea4d","subject":"Blacklist sphinx 3.4.2 because of Flask regressions","status":"MERGED","created":"2021-01-06 17:22:25.000000000","updated":"2021-01-07 04:27:37.000000000","submitted":"2021-01-07 04:26:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769571","meta_rev_id":"856f41708e9f4e945452bf0673f6afc962f332a9","_number":769571,"virtual_id_number":769571,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-07 04:26:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-06 18:12:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-01-06 17:25:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2021-01-06 17:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-06 18:12:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-06 17:23:01.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-01-06 17:34:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-06 18:12:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-06 19:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"011d591320d8181524b87873f62610ce1186bd7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-06 17:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"accc8c33e536971d738d11f065cdd2dc14b54553","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-01-06 17:23:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"78d8829f396210b87fe0d6dd45c46b78cb6a3046","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-06 17:23:04.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5b97317b2c97075c88f678f3f70b64793ecbd5d4","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-01-06 17:25:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"377183883b35b5f801f80f39791aa35308c6831a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-06 17:34:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"587455b70f9a38b3247c98e1da421aeec8accd5f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-06 18:12:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nack\u0027d upstream","accounts_in_message":[],"_revision_number":2},{"id":"b0b468ffb3436e7ccc794ee5bb32ebe427880287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 19:53:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55468ced3af4720960d4f94033f22f2 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c7d21dc05f46b6a612481f53eb076f : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c37a74aacdf40bdac379e9d63b3636d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a3a7134bbd419aa4125883405fd910 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b69ab7c083846278906ff44831a26cc : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/442fec33bdf44573a96b51b86f2475f4 : SUCCESS in 4m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb387b322c104b4097e32e028922d0ca : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8487862ef57b4a4da9ecc2aada781c03 : SUCCESS in 2h 04m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b7281fea639a329b56ce56ece549adf6ad876c97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 19:54:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eebcdd2af63850365dc31cdded17a91072807ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 21:11:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb9017a15ec42fd89ac6f29aca99795 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b31b0afe52476bbd0cd31f1e20c92a : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467bfa86243c4585bffc27117735bf39 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b5eef68a4d34d72b4eec8041f3738b7 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc63600edbec4b97a7357c9b5d8492ba : SUCCESS in 3m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51c22a63fdee497cb35c3720b62aaa62 : SUCCESS in 58m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35e3bc4a88854c4d8f21a1d4e9a0a55f : FAILURE in 1h 12m 17s","accounts_in_message":[],"_revision_number":2},{"id":"bf207a86fa39752922ca2fe152d4ad1ae3a79aac","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-06 22:50:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46ebf86a567c96203663ed62e7336696245ee753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 01:01:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c7d28cd6b5468f87c34a0d97845f82 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7fabe3910984bc79a9bccbb1e960a6e : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808271f6e7ef41fa85951ec0e66f3824 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4b2bb54174347208b03153def4a6721 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d64bdf4e627b497e81a13ed55972fbb4 : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ddc2e48cf534c2991986e9b91e38708 : SUCCESS in 7m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38684f9cb5bc4e899365356d614f16e3 : SUCCESS in 40m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e2aa249cf3148ca89f48b73938c5cfd : TIMED_OUT in 2h 08m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8b42e354b6e2f89a012e14cbf675fdf297cb4a50","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-01-07 01:37:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc4a9153cd4d55c30dda9e1718a2621c58916052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 02:54:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d347d1fb9de94545a6dd387788a444e6 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daaa13f39b6f490a95da152312fd3a2b : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d582d08c84d5413aa71bd33569f75533 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8780e6a6b54f4a659ce72559b84c91cd : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a807e3260e88465ca784949c6cc15d9c : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41e872bb43ec4561a3a9c549cc37bd26 : SUCCESS in 3m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc5202a2bbd7494bb5459043690097f7 : SUCCESS in 55m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9c6a0cdd1a24bada05794b4a07e832e : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b0360966649cada3cacb173f53e45c54e3fb59e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 02:55:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"10862eba213d3d2e5777821cec4c5ef68179e1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:26:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548c7cdc8fd24252a897ca714f5f3463 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e38a3860904d4d9bbb37ea22572214 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54eac9ed77654ecc86943587f8070748 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b291abc35ca34486a72c77fb47b1885b : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3f8010da541f46f49683b855ca0a37c1 : SUCCESS in 3m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89d416134aae4681a3b951e0fe0a035f : SUCCESS in 58m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09e5e06cbdc240a78a0b8f59270b04bd : SUCCESS in 1h 26m 15s","accounts_in_message":[],"_revision_number":2},{"id":"345bfa5548e0dc0df81e10c838a40b81fc991743","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:26:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"856f41708e9f4e945452bf0673f6afc962f332a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 04:27:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f351c033e3354e82835b3a6382561b5e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3","revisions":{"ebc6d3d570cc40ef5c0cc711330cf827d7a72cfd":{"kind":"REWORK","_number":1,"created":"2021-01-06 17:22:25.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/71/769571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/769571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/769571/1"}}},"commit":{"parents":[{"commit":"ff08aee41973835d07a43c1b57363ef2e5e1f7ce","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08aee41973835d07a43c1b57363ef2e5e1f7ce"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-06 17:17:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-06 17:22:16.000000000","tz":0},"subject":"Blacklist sphinx 3.4.2 because of Flask regressions","message":"Blacklist sphinx 3.4.2 because of Flask regressions\n\nUsing sphinx version 3.4.2 with Flask and documentation builds results\nin an error:\n\n  Exception occurred:\n    File \"/opt/stack/keystone/.tox/docs/lib/python3.8/site-packages/flask/globals.py\", line 38, in _lookup_req_object\n      raise RuntimeError(_request_ctx_err_msg)\n  RuntimeError: Working outside of request context.\n\nThis issue has occured a few times with sphinx in the past and has been\nreported upstream [0][1].\n\nThis commit blacklists sphinx 3.4.2 until we get a newer release with a fix.\n\n[0] https://github.com/sphinx-doc/sphinx/issues/8655\n[1] https://github.com/sphinx-doc/sphinx/issues/2796\n\nChange-Id: Ie80e45571efafe73a553ddd89c5be90fe97bea4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebc6d3d570cc40ef5c0cc711330cf827d7a72cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ebc6d3d570cc40ef5c0cc711330cf827d7a72cfd"}]},"branch":"refs/heads/master"},"8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-06 17:23:04.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/71/769571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/769571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/769571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/769571/2"}}},"commit":{"parents":[{"commit":"ff08aee41973835d07a43c1b57363ef2e5e1f7ce","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08aee41973835d07a43c1b57363ef2e5e1f7ce"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-06 17:17:56.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-01-06 17:22:46.000000000","tz":0},"subject":"Blacklist sphinx 3.4.2 because of Flask regressions","message":"Blacklist sphinx 3.4.2 because of Flask regressions\n\nUsing sphinx version 3.4.2 with Flask and documentation builds results\nin an error:\n\n  Exception occurred:\n    File \"/opt/stack/keystone/.tox/docs/lib/python3.8/site-packages/flask/globals.py\", line 38, in _lookup_req_object\n      raise RuntimeError(_request_ctx_err_msg)\n  RuntimeError: Working outside of request context.\n\nThis issue has occured a few times with sphinx in the past and has been\nreported upstream [0][1].\n\nThis commit blacklists sphinx 3.4.2 until we get a newer release with a fix.\n\n[0] https://github.com/sphinx-doc/sphinx/issues/8655\n[1] https://github.com/sphinx-doc/sphinx/issues/2796\n\nPartial-Bug: 1910419\n\nChange-Id: Ie80e45571efafe73a553ddd89c5be90fe97bea4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
