)]}'
{"id":"openstack%2Frequirements~769664","triplet_id":"openstack%2Frequirements~master~Ic0edce86ab63238dea7f99b2790fc527c6e080ef","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic0edce86ab63238dea7f99b2790fc527c6e080ef","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-07 06:21:00.000000000","updated":"2021-01-08 02:38:01.000000000","submitted":"2021-01-08 02:36:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"769664-openstack/requirements/constraints/noclob","meta_rev_id":"8109aea826155242372d0c8ffff1906551e10786","_number":769664,"virtual_id_number":769664,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-08 02:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-07 06:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-07 06:36:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-07 06:36:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-07 07:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-07 08:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-07 14:42:41.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"CC"}],"messages":[{"id":"e218c2dd409ff842ec18f468ddb08ec9eed601a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-07 06:21:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56b868dd363dbf274c37fa6d8b606e4c304ccc82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-07 06:36:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"154d014003e05f74568d685171d3ed0b5ddd6bdb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-07 06:36:17.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"541ef7b6a7a56c1706537dccd5d59975c14a30bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-07 06:36:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7599c490341f5b04f0190b2e158fc389fb6aa16a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 07:08:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/edf4b1d40ca64204b07c47b46e76d647 : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":2},{"id":"068fdf2871cc78a98b24bde96770b0da032d1ec1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 08:13:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598d34ce69914a7689a5fe648b0095b7 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c95c307b09412a82682c20b1452013 : RETRY_LIMIT in 2m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dfad651498d4886881031c29c5cc522 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7d7f946a50d4df39a19df6fb5f12068 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cec92df0d36747559a69211022f5e6e8 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7382ec6c00e46f5b7a8824ee3d493a2 : SUCCESS in 4m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53af043538364091bdcfcfea66c88b70 : SUCCESS in 4m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/69afdc0ecdd740baafd9cfda6d31389c : SUCCESS in 9m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/16c9a7e523e24455a3d2d8e5effd9f8f : SUCCESS in 8m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd8274bcde874b1291df779785f9c7bf : SUCCESS in 10m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0878e27042df45cf9004cf3a7de264d1 : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0092888d923a4e14821a5ad341cfa6c9 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ed3e2cb452704bd8a62acb60ca6f51d8 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6101218a4fd7485d9ed54a6ac39ab58e : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f381ab5f323c49dcaa00868afddea882 : SUCCESS in 12m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99ec791ac11e40398cef53a44caa1047 : SUCCESS in 33m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/61d2ea48ebf543e08e1f2ab1c6804d41 : SUCCESS in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2fa9e3e489d4480ab2856316d829f999 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc043e81ab74496eb525587a5e8d2516 : SUCCESS in 19m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b56c818c09444c7784e9f4476e2acfb8 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ee0760c9a004b9c84bb3aba580d6473 : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c269633af75e43bda923d7a46f3cb0c9 : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4abd2658d95b4ca18445409e0a1ff25e : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57fa1ac8377746bf960487537c5698e5 : RETRY_LIMIT in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"8426e740ded15106094a7cef496e4c8b9897dda2","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-01-07 14:42:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c38af555c23d4537e32e20435fabad150bb5ee28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 15:12:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9d304958f5964ce7880030a37d0c415e : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":2},{"id":"7cde58de98e830b974e7fe67f30e013636c5a988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 21:09:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb88ba4fe824ec79f9497ada75b5cc6 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780dab2b59b34677afe2192cad67b42e : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9805023f92994396865f2ab2f9075e6d : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1d560c561ba468d97497a2614780a29 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71ddbf73c2db41038ebc122ef013f779 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7eed71d2ddde49a596bfe7fe7f3cc6d0 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdc51298e8884216a459052158a7d074 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b821962e8a740fbb80107e42ec04fce : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d35059e7b3c4e3c9275efad908dfacc : SUCCESS in 22m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2c948ba4558648a488372d7ea984f9be : SUCCESS in 8m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9705d54c5b9049f1ad359eb77b82d92d : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4e004e6c46fe467582c7899923c5dab4 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7616d61abaa412686463436d3c17247 : SUCCESS in 10m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ed2e0d61ecca40a39fbe6c86d95e6493 : SUCCESS in 8m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/333802a8e1484bd4987337fbf186e0e2 : SUCCESS in 15m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/74ca1de6001d43949a9059f9833481d2 : SUCCESS in 36m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2b9a55cfd7144403894c9d58a379c6e9 : SUCCESS in 13m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d40208d0e85d4a789120df1b80e78711 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6c01c1f154f40ac9fc5943eeb803479 : SUCCESS in 18m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a76cf91db949452781230818e93a8b20 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/133347c7b5914608a749de08aba8472e : SUCCESS in 6m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5dd9afc1c47746008a2e270c333aa362 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c165757f75d1422499d873960777cc2c : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/503002b23e7c4403849597d196df486c : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2a00ed96f7ebef487c152fddb0aac0ab9377e83d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-07 21:10:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3265fea0470f11524e3ef1bbe6462f76ac3c36b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:36:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92520731c95d44a7bf0d3b6d61e39049 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38694cfa5e5e48c1bd6ae446546e7efa : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26b53accd4cc44d0b262ce989ea10520 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79711700979946b1ab3e1220ce6e7094 : SUCCESS in 4m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9aaf05f08ac4ab496ec9fd760b42660 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8e3481b5bff740dab6445b81cf124da1 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d75e945c8a0c4212b32bf1d80ec75ae8 : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ee32ef84a7d94efc991d86aada6566b2 : SUCCESS in 7m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b79065a2310e4fe78b24fb714308b3a9 : SUCCESS in 8m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4f3b5bf2ab9a4f4b81bfe4d57c888045 : SUCCESS in 5m 46s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0cdbea03d5a048dd834022dcf62ef391 : SUCCESS in 9m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae2c37e9c5814762883c0b1bd85076d6 : SUCCESS in 6m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f898d3d9469a436783d98337176bd944 : SUCCESS in 8m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/88fea8d07629443ba190cdc0f8e6a42e : SUCCESS in 14m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/83b6fa4ef5f2414cafb809c63a9ec7c5 : SUCCESS in 31m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0e09ab312b864614a247bbd41dd84756 : SUCCESS in 12m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/be0ba0ae3a1748ee9ed17fd38cff7870 : SUCCESS in 4m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5b6231c1f0074dd9bc6e68b55bbac48a : SUCCESS in 20m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fdd13ab7020a4f74b3592db8e328bd1d : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0447b63b29c94eaab67cd7c5d305d2f3 : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/00e58b23c4b94ca8a1e441c4973dc3a0 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea3ffd6b3d7d4cb9947008780b7885a4 : SUCCESS in 57m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da109f55258247599328f9d94b1b1081 : SUCCESS in 1h 16m 31s","accounts_in_message":[],"_revision_number":2},{"id":"24b1f40afb250a0f960a751b329e8b3cfcff6cfc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:36:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8109aea826155242372d0c8ffff1906551e10786","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 02:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a0a3ca731d4d689b04434d43b7dc9f : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"597caa49471f8572377d7ceaa32f5c249ebed138","revisions":{"3329bebea34a7d53b23dc558f474974593db5a7b":{"kind":"REWORK","_number":1,"created":"2021-01-07 06:21:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/769664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/769664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/769664/1"}}},"commit":{"parents":[{"commit":"8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3","subject":"Blacklist sphinx 3.4.2 because of Flask regressions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-07 06:20:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-07 06:20:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic0edce86ab63238dea7f99b2790fc527c6e080ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3329bebea34a7d53b23dc558f474974593db5a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3329bebea34a7d53b23dc558f474974593db5a7b"}]},"branch":"refs/heads/master"},"597caa49471f8572377d7ceaa32f5c249ebed138":{"kind":"REWORK","_number":2,"created":"2021-01-07 06:36:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/64/769664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/769664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/769664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/769664/2"}}},"commit":{"parents":[{"commit":"8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3","subject":"Blacklist sphinx 3.4.2 because of Flask regressions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b7a53cac1cc1fdcd4ab94aad9a7b7d24362fdc3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-07 06:20:57.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-07 06:36:06.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic0edce86ab63238dea7f99b2790fc527c6e080ef\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/597caa49471f8572377d7ceaa32f5c249ebed138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/597caa49471f8572377d7ceaa32f5c249ebed138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
