)]}'
{"id":"openstack%2Frequirements~769875","triplet_id":"openstack%2Frequirements~master~I958ff99870eccef9b264904599971c7884d9bb6f","project":"openstack/requirements","branch":"master","topic":"drop-horizon-blacklist","hashtags":[],"change_id":"I958ff99870eccef9b264904599971c7884d9bb6f","subject":"Drop horizon from blacklist","status":"MERGED","created":"2021-01-08 07:13:44.000000000","updated":"2021-01-15 21:48:47.000000000","submitted":"2021-01-15 21:46:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"769875-drop-horizon-blacklist","meta_rev_id":"2a6dc9e7329cc580ca1c33663b1f1fdc6d8592fa","_number":769875,"virtual_id_number":769875,"owner":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-15 21:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-15 15:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-15 11:03:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-15 15:17:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-08 08:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-08 17:00:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-08 17:01:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-09 15:08:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-01-15 11:03:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"a4db4181bfeea252c6cef6d99124a636b056e214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-08 07:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9aa7851f86bda7dc6527f147def18c6ae136c42e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-08 08:20:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d101851bfc54413fa036781fd46d953b : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fe6e7f76b784d9288e120ef7342ed66 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43032a602af24daabfb0a4897b1aeb64 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e605f6d14cd54ffbb924c085c3c60b4e : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f3150a4449b446daf05b16b52b177e3 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/237efd52fa8a41a0a2689aca94c25ce7 : FAILURE in 4m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3775cbccdf5b471aa6c7abf1450c5403 : SUCCESS in 55m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea3e2ede2c33477dbcf1dedc1afc98db : FAILURE in 1h 05m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c2f03eac1f7667fbab8c15c5baab92516fca00df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-08 17:00:10.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nI confirm the previous cycle model\u0027 switching \nhttps://opendev.org/openstack/releases/commit/427f5952cc481f11d3e81e2ac039799a4ca5fcc9\n\nHowever I\u0027m not sure of all the impacts of the current changes, AFAIK they LGTM and they don\u0027t ring any bell on my side for now.\n\nI only +1 to allow to grab more feedback for now.","accounts_in_message":[],"_revision_number":1},{"id":"49eeec0287eba5315edf46e20d95a1be7030a485","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-08 17:01:39.000000000","message":"Patch Set 1: Code-Review+2\n\ncool, I think for now dropping from the blacklist is all that needs done","accounts_in_message":[],"_revision_number":1},{"id":"24b6d85b72031a2535e2366e722c231f5f62e7a1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-09 15:08:11.000000000","message":"Patch Set 1:\n\nNeeds to be added to upper-constraints.txt yet.","accounts_in_message":[],"_revision_number":1},{"id":"f15eaed50ef5311b419749c1bbc4bc7a35c3b680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-13 14:14:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d6d1fd7e41cfb364adba550f77b508a37dbb359","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-13 14:32:19.000000000","message":"Patch Set 2:\n\nWe already did the same thing for some stable branches (at least for rocky, stein, train) to avoid gate breakage. (We did not do it for stable/ussuri and stable/victoria so far but there is no specific reason. I just forgot to do the same thing.) Thus, I think there is no side effect of this change.","accounts_in_message":[],"_revision_number":2},{"id":"f4d114f1c64a1b132337cd1018d48bec9c43def2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:38:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/51fd9d75ca6d4444b4a467f5e60ee613 : RETRY_LIMIT in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7cff44558c1e2db1899907589321ad69401ca1b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 03:51:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85982e7dc5f24dddaa7023071bdace6d : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97328d6316da4f54ab1d5bf4de6b7efe : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9742d30b281d49518e7cda3e8469ac17 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d94c393c201d47d4b819f8529e21a18f : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/820cdff0e90e491ba4555d76dd283d35 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/63a56dd2171e4ac595ce14c730c536f0 : FAILURE in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0452d58bec5a4f0fbe261783852edf40 : SUCCESS in 3m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ee1fce4695224670817a69039dfff06b : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/478099f79b6b4c9c8003fe6321afe3c8 : SUCCESS in 15m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba953c5f8b0941f89adb892b94390e44 : SUCCESS in 11m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bcfc2d39d1f847bf8533af0c90eb88e7 : SUCCESS in 7m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b773c02c45fb4fb2857ccec2930af646 : SUCCESS in 10m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9459434443864cbb880db77e49885b8c : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3894d0a3e82a4a5fae1eaf9f74ec39a2 : SUCCESS in 6m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8c2f84cf336a487fb6c194fbec4e7e34 : SUCCESS in 41m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7ed35a8a87c844e29fb52ee37035ba25 : TIMED_OUT in 42m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/68967afdf2344f619efdb3dbda7f8338 : SUCCESS in 13m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9c1b4649163f46b9b08bd1e373c1b3ba : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d5bca72a426a49b7b58a53eebf9cd1d6 : SUCCESS in 18m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/412ee66fb7954c6586daa6df769e9a3d : SUCCESS in 4m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5364e28eae6e4dc7981e1a568ae357da : SUCCESS in 7m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6cc4e7e724ff4d568e83c8c69945a25d : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/871fad3f9d6b499cbc63900df14871f4 : SUCCESS in 53m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df2410bf1dee44bcb112d7cb58a124dd : SUCCESS in 1h 30m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3759a3d5ad106e5acf1cc41c153e76bc26f49a47","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-14 03:59:20.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"00aaecbe1e558a37268e3d6617ad0c2095b131b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 04:14:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6ec3dc2e48584148962082eee8f6e71b : SUCCESS in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"93bddc33cf809e7192b52381dc14e5ff96bb7d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 01:15:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f997f518d6410793a65536836708b2 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e16dea1cf6f4bafb27bc2807fa0448b : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd07a410b774435ea019962e739003df : SUCCESS in 17m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1073c24872784c22a1653d4b3b1d6dc6 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cbf8129e8ec41e8bbb837a5edaaaa5e : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab339c517aff4c1bae2e3c3ad175a59e : FAILURE in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e7735e1dc1d0439ca43218dcdcb2de01 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f9361195c1d4b29a7496c64d3c216d1 : SUCCESS in 27m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/13d41620609e4e7fa6ffa608e14a0d63 : SUCCESS in 11m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/bcd4110c73894cbf91bd52d6ee5fdf1a : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a4866b3d45b74d5ebad6235022c32618 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f62e3f92aa2442989405c65e5d0f880 : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c81ad58c8a545b996d288bec4e1102f : SUCCESS in 6m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e480bdc5344a476cb6be154163c6cccd : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a34916c10fe247849d2055e8fa258b04 : SUCCESS in 17m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f12528efac74c2d82ff17a2ae4e85dc : SUCCESS in 33m 07s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/48ac237d5c4a45d4b3f030b492ed4459 : SUCCESS in 29m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dbcd08ffb6664d4581a902e57e1dedc8 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc31421b3d9449adacf6e517137035de : SUCCESS in 16m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/932b6fa7e46b492f918256abbf08c77e : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0598e8e1500b4f7294c53f494453b99e : SUCCESS in 7m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b5d945c228c4450d9bd061861dc5916e : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3699d3877ad1451abfeade9a92035abe : SUCCESS in 52m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/966097dac6764ce2ba4e04c07381d84e : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d5bb5d5bd2b3c0bbd809e4c7565de092b7ab7bea","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-15 07:18:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"647d39939814aa0c080365d8527f50047b0e4c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2021-01-15 07:19:39.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"720b354dba9d9641264755bc1df42f970d4e1c60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 07:40:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5346c1fd12964755b2d381f8cf6bb95b : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4b9c6390df0f9d71204e8e0afafc519b55530e9a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-15 11:03:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"84f1e59d61e217a4061b376c086d0715a5c3776e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-15 15:17:18.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e0b2fcf1fa70fb19bde5ffa40500fdec113df253","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-15 15:17:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e515bffbfc4638328c272ef743339f14e4ce7239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:40:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97e973ea0bc425bb0ed828672212baa : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0038b4c07316415b812ff2bb37f226b9 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6189763618e349619e4ac4205c6c4763 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c09ad5d021e8418baf07ed256208bfb8 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ab0d1c252d8466cb0e9c82f3c9dfc42 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41430b9df39949b68391591bd2801a71 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4403611ad13494481cb0d9b634b2a62 : SUCCESS in 3m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36fa3c074d394c7bb5adde2209a841f0 : SUCCESS in 9m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f3cbc925a41434ab62e869fb2424b6b : SUCCESS in 10m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/15c6305e28ab4635bb9d356a9d58b1bf : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/980e922778464ecb813163f785c8237f : SUCCESS in 5m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81580b446ea44eacb6d7b751e3123679 : SUCCESS in 8m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e3fe214ca574bc791f5e4cccbda6d11 : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0c4b91df250449e5b4c7fb3535918795 : SUCCESS in 6m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c54889a5e794289b95dfca4cf199371 : SUCCESS in 12m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a7384338a471404196f5422de334ff47 : SUCCESS in 32m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/23ce4515caea47ff846460092d98dfec : SUCCESS in 16m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f922f34b13f400d9d02d6813f3ce769 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba416a2e897f45e1b4590df902bd18fb : SUCCESS in 19m 08s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b5dcf125e7f3413788fcd5f09c439989 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/54a3c1216b4642c38e31c86d73df25b2 : SUCCESS in 6m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e2c5c0c31d8242699a9cb15f5809ed2a : SUCCESS in 11m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98060a656eda4d23856a69603395ed9c : SUCCESS in 56m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3143b7fee0414226a530a1c050049d24 : SUCCESS in 1h 25m 25s","accounts_in_message":[],"_revision_number":3},{"id":"8bde0f4e64ee6ed2d804084f6d0115856c22530b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 19:41:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8df858c58f6b5ae53107bb08f275ebd4337272a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:46:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0fbabeea36946c4b6c89c72514f5ac0 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bfc8ed452b74b0baea8079002829e17 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b8321dff4514291be7475ecc0090027 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13c16a025df481987f45d0ecd5c9ad1 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9741d1bd69a943c7b66ead35de024b4c : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb4e0f74dd27451f840a2b11eb030287 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b66c1d41f2f247ada3907f772417d261 : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f692efb7f654425d93824c77136ca337 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/06dd33bc80da402fba385bc6081167ff : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/41efeb688e744c748b9e5e3bf7794d3b : SUCCESS in 5m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/226e0dafdf2f483a889fdaab5245e360 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/629580336475488b94f0b831f58441ff : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21b1dda9e5544f67bb5eb738e258ee3b : SUCCESS in 7m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0a6a3e78ef14f998239aabd428d1f2b : SUCCESS in 51m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0f110e64fa3749469cc51b1cc09e8428 : SUCCESS in 26m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/448ac2b8b88b40acb0b6d543a5b2113e : SUCCESS in 12m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0e234ab14d254dc6a1c0445b5cb0fa23 : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7544bc42d8942dc9235a0acaa7d8d1f : SUCCESS in 17m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/9d64fb27e6d441cb9eda5b62a2bdb420 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a1ece5eaa7345c8af5d67a3a37a66ac : SUCCESS in 7m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c1d4b11c290646e1818d9060a7f6ef3a : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ceaa2996ed84d72bf92feb4df884d34 : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b6ec95aaf4b47458006a20c0641c5db : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":3},{"id":"1752d46a0dc8a6fb706d093f2e158f240f7413d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:46:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2a6dc9e7329cc580ca1c33663b1f1fdc6d8592fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 21:48:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f989ef36d624dcfbeeb8bb1975cc3d6 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bafda3dca1dccdb88e432eb1c8f557c9630ea67","revisions":{"8a78b20928ddee039d6285bf5c74fa17af578038":{"kind":"REWORK","_number":1,"created":"2021-01-08 07:13:44.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/75/769875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/769875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/769875/1"}}},"commit":{"parents":[{"commit":"ff08aee41973835d07a43c1b57363ef2e5e1f7ce","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff08aee41973835d07a43c1b57363ef2e5e1f7ce"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:14:55.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:14:55.000000000","tz":540},"subject":"Drop horizon from blacklist","message":"Drop horizon from blacklist\n\nhorizon switched the release model to cycle-with-intermediary\nin Ussuri release so that horizon works as a library for horizon plugins.\nI think we can drop horizon from the blacklist.\n\nChange-Id: I958ff99870eccef9b264904599971c7884d9bb6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a78b20928ddee039d6285bf5c74fa17af578038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8a78b20928ddee039d6285bf5c74fa17af578038"}]},"branch":"refs/heads/master"},"be4aee97ed6634235f079c63061b9fc66ef4767d":{"kind":"REWORK","_number":2,"created":"2021-01-13 14:14:31.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/75/769875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/769875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/769875/2"}}},"commit":{"parents":[{"commit":"91784cd3bd5071f2a1e2efe24c512d75a1a70b91","subject":"update constraint for neutron-lib to new release 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91784cd3bd5071f2a1e2efe24c512d75a1a70b91"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:14:55.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-13 13:52:58.000000000","tz":540},"subject":"Drop horizon from blacklist","message":"Drop horizon from blacklist\n\nhorizon switched the release model to cycle-with-intermediary\nin Ussuri release so that horizon works as a library for horizon plugins.\nI think we can drop horizon from the blacklist.\n\nChange-Id: I958ff99870eccef9b264904599971c7884d9bb6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4aee97ed6634235f079c63061b9fc66ef4767d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be4aee97ed6634235f079c63061b9fc66ef4767d"}]},"branch":"refs/heads/master"},"7bafda3dca1dccdb88e432eb1c8f557c9630ea67":{"kind":"REWORK","_number":3,"created":"2021-01-15 07:19:39.000000000","uploader":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"ref":"refs/changes/75/769875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/769875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/769875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/769875/3"}}},"commit":{"parents":[{"commit":"91784cd3bd5071f2a1e2efe24c512d75a1a70b91","subject":"update constraint for neutron-lib to new release 2.8.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/91784cd3bd5071f2a1e2efe24c512d75a1a70b91"}]}],"author":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-08 07:14:55.000000000","tz":540},"committer":{"name":"Akihiro Motoki","email":"amotoki@gmail.com","date":"2021-01-15 07:19:33.000000000","tz":0},"subject":"Drop horizon from blacklist","message":"Drop horizon from blacklist\n\nhorizon switched the release model to cycle-with-intermediary\nin Ussuri release so that horizon works as a library for horizon plugins.\nI think we can drop horizon from the blacklist.\n\nChange-Id: I958ff99870eccef9b264904599971c7884d9bb6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bafda3dca1dccdb88e432eb1c8f557c9630ea67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bafda3dca1dccdb88e432eb1c8f557c9630ea67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
