)]}'
{"id":"openstack%2Frequirements~770052","triplet_id":"openstack%2Frequirements~master~I0b2a044f8331fbade632adf9bf17dfb451714880","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I0b2a044f8331fbade632adf9bf17dfb451714880","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-01-10 06:20:41.000000000","updated":"2021-01-17 20:52:08.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dabfa2884dd4a22f51ef4a306132ef54b5e99589","_number":770052,"virtual_id_number":770052,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-01-11 11:03:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-11 09:33:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-11 09:33:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-10 06:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-10 07:18:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-11 09:32:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2b68e5e93d91a8fb3eaacd33d0c1afa7c2ea6e57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-10 06:20:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecb3faa2902f6b98a935d9feed036d031626ad24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 06:41:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/287a502803d34f9ba312990470e60bb2 : FAILURE in 14m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dedf8610bbdc6400e2837b1e1590403de0f17c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-10 07:18:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1b0dcb6ca94ab68fc79e256d5503bf : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a28bcf2d8b4d19892df18e7336aa24 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ed688e28ed1476ea422d7e0f1e7192a : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00086d75698b481dbd32152ad312ca3d : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d50c43122d3409d94235925e8ba1311 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/852a083697044f08b2c7347e6c10247e : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4c93289a6224f7b8012e634177704b3 : SUCCESS in 6m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/28bcee9470f949db93a61d973a55f2d0 : FAILURE in 6m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e548afb956de429d8f808d8e63ff29e5 : FAILURE in 5m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3c51bb5cf014a8baf8f5bc432be3207 : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/af0a19173b194aebbe6c1941821c4857 : SUCCESS in 6m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1c730b1d183345cc9a65a71f73b36c6c : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ddee8d23b4cb47bdba33e7dd7b631f1d : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7adf605f0fef44058632dafa7a6ffbf8 : SUCCESS in 7m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/689a989e6d6b48e0967b638803a8da30 : FAILURE in 12m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a5ed59086b714273a226ee10b2a2fa87 : SUCCESS in 31m 47s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e015657f49e0462ba845aeb7ed3f7984 : FAILURE in 13m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29a22692f2704cf28aeb635470ad15e2 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/57fa2af9126d48bba1faab908b86e5a6 : FAILURE in 15m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f73d1b6d14cd4f4c85d228ca9221f7af : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2ea57ceb90f4083a7141723c7d8f65f : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bec915110ce44b7eb7c074aaeab08b7e : FAILURE in 13m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05dd8585264847c0bd69cb62dbf39bc5 : SUCCESS in 54m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da4857f449a840d9a242153b5c41faf6 : FAILURE in 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"54b9e3c4df43c76e8b760b1cc28a06cbcdb18ee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-11 06:31:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"445d46451a12c5a13e3c0789d4329a38f5aa3e3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 06:47:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/71241354fe37420fbc64425eed487901 : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0474a161cfcf3dcbd0ed122c9f789fa46ae17df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 07:54:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c65d25df554299a61f62ae016d8e27 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20eb8241592e4fc6a94a22974ca955ac : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8741a40cf9441ab9eb1c5c48f8bbb97 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34f92040423f462690b62578ca71350c : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b9b17b48014494a9493d24d2df809a2 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7277da24d1b04d6bb1ac3063c22d4d2d : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3a091004d3fd415cb55a45178593b015 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6083ff3ee0304262bbf25a645b566c10 : FAILURE in 4m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72083be48ff34ae685e0db236a367637 : FAILURE in 8m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/77fdd479cc9847ed99b5b96add1cdc54 : SUCCESS in 7m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee11f60b105e45df85808cfc22df73d7 : SUCCESS in 6m 01s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b8450aee5fa04ebeab1f559329bc26d8 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/051020c6f3144c8a9f65b7e10deda1d8 : SUCCESS in 7m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3dee0abfb9048fdbe877929b4c00c57 : SUCCESS in 6m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6eba0e62d90c4497b0778f7af80d988b : FAILURE in 15m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c01546206f754796be5b3df8515abe80 : SUCCESS in 37m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/16b447d5d697457a929735abdb17c11b : FAILURE in 14m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/647add5406044651adb6d1195f07d3b1 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c89c7fd42bb475684a34d7326ad60ff : SUCCESS in 19m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a58bdae57d8d4289a93e66f484e4cf72 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4dbe21420fd4f319f09179012d6ed57 : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7feac02875814eb69f93a2923db71fe4 : FAILURE in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd12d8eb05684e9aaacd7a17b990e631 : SUCCESS in 1h 06m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/463f8230b82d417f9e80287e16452fbf : FAILURE in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"341c57ca27a373eba7e6b541019f03965149ce47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-11 09:32:27.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"36a835df6743e5128587c6ce5f37f466a4655652","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-11 09:32:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e601dddddb6655c617af9612e9e36e0c61e66d70","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-11 09:33:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"43041484ae36bd81b2fb5d0eec002f9491fc81fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 10:05:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7445f59f2774bcba598a4afd3e6eb66 : SUCCESS in 27m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ab2fef526ba8f8aa2e71553a094714dbbf02f07c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 11:03:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49ae9f7fd9540d6b67e34970267b59f : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5819ef04c89f43d3b221f2cca463105c : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e565910f5b014739b2fa81f6fd4d1dd1 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f19a3d7f65fe452ba9882d553f5893c3 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09cb4e4f2f6d499dae5e9b4b04e7c0dc : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1b2c0033d76462b91ed9cff48063b5e : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ddb94d1c975048e188beb514c543926c : SUCCESS in 5m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f0b8f15ab5d1420e8519f54df88f308e : SUCCESS in 9m 56s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a869d2b0d2344afb9e9ba47825b5319 : SUCCESS in 8m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/877a1d6d353a43c3a29cd78f4ee0c24d : SUCCESS in 8m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea91590c3ba3440a949afe2f04a1a6f8 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ae5f4efc25f84ece9947637d66c10d0c : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/381d1d1216de4b3784a48609de0894ea : SUCCESS in 6m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3cdf48734e0a43f7b419903e6a614a32 : SUCCESS in 7m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39053905348e490ba2b102e41919f7d4 : SUCCESS in 21m 59s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6da98e6ac0e4a40b0860ea2ec4b7d23 : SUCCESS in 34m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2878032c670f49678fd3cf91cf417f5e : SUCCESS in 13m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5214d261dc01437a8d2fae3651b55cf0 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17a6ec08f9494f869f61a5cfc26aa0b9 : SUCCESS in 17m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fc39c26393a149edab08627c501ab97e : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/db82336472eb462faefde1ac931da8a3 : SUCCESS in 7m 06s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dc66ae10902f41fa9a23f22bcc6cadb5 : SUCCESS in 10m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07edf64c318a4f2e9db70b74ec5d6efa : SUCCESS in 51m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c69e02e5f0544f36b55253f99f04d869 : FAILURE in 20m 22s","accounts_in_message":[],"_revision_number":3},{"id":"1ae19d66d4d0c211c5fa6422f336e20ddfaad839","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-11 15:15:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"901eaf94ed7d4472500546976f5bdc5a93687d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 15:36:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8f4e76240ca04546a9f432fcccfe305d : RETRY_LIMIT in 2m 40s","accounts_in_message":[],"_revision_number":3},{"id":"a20857677f79428b94b081e671f7e8bc7a94ca49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-11 18:39:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9361a95ff714498da2a6c698c9da4ada : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13d7df38b9ce4d17bef0c075e82f4495 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34b9e64ea0394259848aaa6a94f90e59 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6ee46635d8b46688b1d9f51f09f2542 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32d8f1d1ecb64109af9d15c645be15a3 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98fe4fea188c49d1bf3af18d4f54c745 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50ab53bd4f8d49d4a8cbc12f2d60c107 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7ff926812d0f4728a8b20a0dfd564a34 : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6dd9bf74aabb4d029c391a4646c52920 : SUCCESS in 9m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3e29449fb05b4ec3af2c99824d87464e : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d002ce79a8d647c48241861a21c6be7a : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bcd8cb7378734653bfba06558e0a7908 : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/035b14305e71403db0ed76deb29f5c33 : SUCCESS in 6m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a83dad76738d48d980d15e61bfdb7228 : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1e483e951090468d9933f1b2f90f3c0d : SUCCESS in 12m 49s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8e3fd83fa6e48d1a07c7bf5f0e8ea99 : SUCCESS in 31m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fcd3fdea899c4811ac72548220d1c9af : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b47dc2fcf45547faa7fe1160b2b2ff0c : SUCCESS in 4m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3feb7a464f084dbcbda78585a651908b : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2bd9f135a78f4fe29774b05e288a9b46 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b32a3ac85cc349f98058c1d03126e20c : SUCCESS in 6m 21s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2c75d58a7dc34784b965d76f57c295e9 : SUCCESS in 11m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6efd90b9b444006876a2d233bed83b9 : SUCCESS in 1h 08m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72d127a7e84a4e9db722b3a28a0a5b13 : FAILURE in 21m 07s","accounts_in_message":[],"_revision_number":3},{"id":"dabfa2884dd4a22f51ef4a306132ef54b5e99589","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-17 20:52:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7046c4be65b979f035f8268fd66c204978828262","revisions":{"0a25a6942974a38674c0097c605f3083f061f4bd":{"kind":"REWORK","_number":1,"created":"2021-01-10 06:20:41.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/770052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/770052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/770052/1"}}},"commit":{"parents":[{"commit":"b767e3e11f442b12f29fbcc61ba5188353f8f4f4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b767e3e11f442b12f29fbcc61ba5188353f8f4f4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-10 06:20:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-10 06:20:40.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2a044f8331fbade632adf9bf17dfb451714880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a25a6942974a38674c0097c605f3083f061f4bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0a25a6942974a38674c0097c605f3083f061f4bd"}]},"branch":"refs/heads/master"},"e9508dddcc3c3541be9c52b7405381d2657a7ea1":{"kind":"REWORK","_number":2,"created":"2021-01-11 06:31:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/770052/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/770052/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/770052/2"}}},"commit":{"parents":[{"commit":"b767e3e11f442b12f29fbcc61ba5188353f8f4f4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b767e3e11f442b12f29fbcc61ba5188353f8f4f4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-11 06:31:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-11 06:31:31.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2a044f8331fbade632adf9bf17dfb451714880\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9508dddcc3c3541be9c52b7405381d2657a7ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e9508dddcc3c3541be9c52b7405381d2657a7ea1"}]},"branch":"refs/heads/master"},"7046c4be65b979f035f8268fd66c204978828262":{"kind":"REWORK","_number":3,"created":"2021-01-11 09:32:52.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/52/770052/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/770052/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/770052/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/770052/3"}}},"commit":{"parents":[{"commit":"b767e3e11f442b12f29fbcc61ba5188353f8f4f4","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b767e3e11f442b12f29fbcc61ba5188353f8f4f4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-11 06:31:31.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-11 09:22:07.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I0b2a044f8331fbade632adf9bf17dfb451714880\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7046c4be65b979f035f8268fd66c204978828262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7046c4be65b979f035f8268fd66c204978828262"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
