)]}'
{"id":"openstack%2Frequirements~771192","triplet_id":"openstack%2Frequirements~master~If72068839df8fc39cb6f923f3e914f6db5134210","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If72068839df8fc39cb6f923f3e914f6db5134210","subject":"update constraint for oslo.messaging to new release 12.6.1","status":"MERGED","created":"2021-01-18 12:19:31.000000000","updated":"2021-01-18 18:43:17.000000000","submitted":"2021-01-18 18:37:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771192-new-release","meta_rev_id":"38f7f197725801418fb13238327ddbf608af0cb5","_number":771192,"virtual_id_number":771192,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 18:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-18 12:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-18 15:11:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-18 18:43:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-18 15:11:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 12:28:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-18 13:11:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-18 15:11:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-18 16:17:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-18 18:43:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"54f8fd6a3eafaad53fc64b2234f0d8a8353a374d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-18 12:19:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1868b1d0407aaa39a7999744aab95ad1e45d219e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-18 12:28:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a7cf0ab91e14de8a2bc622e6476bd1f648e6444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 13:11:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2ff95e85f836416aa06e565a45a1621d : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8b5138b4857f6832597315752c26bcba9aa31b4a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-18 15:11:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cb8a6a9ddd505e4f1f7fbe8319a77db174b9f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:17:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6ba29bf5fd46bcb7f656b242491dd9 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd8cd05549d14733a873a6e16941228e : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0e66ece497a4bc4937987141af4502a : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2bc0e2b6c434bddbbdaa45b5ef033fb : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9395f1d25909414d958551701d63f4b2 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f7fc3d97614b4e1ca914393fd22bed6d : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b1e305f09ae64b8ba56aea657b18c7e2 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6720967f4a514ccd8ad25d0374b8f6d2 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f75729c5ec4c47bd9b8702d342b098f0 : SUCCESS in 10m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40d226bcd72e487fa2603a4cf070e52b : SUCCESS in 10m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df09c19fbd454f14af89daacd9a26191 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ba5d98db50d940909c738176d863461b : SUCCESS in 13m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6ec2f1809534d6aa6fc23b67085a7f7 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dbff730504ee4e0c84fac13c7f81172f : SUCCESS in 8m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b692c9181c3f4cc585db2cb84bb9c0ff : SUCCESS in 14m 16s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a28f89b39e6549f3b251851b25abe6e2 : SUCCESS in 26m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/35d98f9e99d948cf8f44d1652fddad17 : SUCCESS in 15m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f10987fb9664f12af99b3f1601d8552 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eaa8c475f5f647bc9fff173d9cc7c71b : SUCCESS in 19m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0ae695444a864512b3a6a147259fe26b : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/595ca1bb2bd64ef2b9aaa87dc6979441 : SUCCESS in 6m 57s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1c5a60d8a0fa40e0aaec9cd82f9ba312 : SUCCESS in 12m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55219b5adc0b4db19c5c0806860e4156 : SUCCESS in 1h 11m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc692b49f4b544e0a17ad6624eedba2a : SUCCESS in 1h 46m 18s","accounts_in_message":[],"_revision_number":1},{"id":"98cba578f4c892552c47ae74cc2c76afd3baf8f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:22:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f24962fede74c9358fae7230bee97994aeb99f72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:37:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0549bea591034a7d83053ab2aa3fae97 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2355eadf9946c59991d663318e6a58 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d50b68dc008b4aa49ce59538d0a6095d : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5959ac984a4f43c8bdc1a4bd1871382c : SUCCESS in 8m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dce1715c5b2d4822bdd8f020dfe12ce9 : SUCCESS in 6m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d545f9d5ecdc4b1cba9f1da0d910d16a : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/314e06b95da040459ce06192267dbb5c : SUCCESS in 14m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9a4a8a0a72864ec0a1b54e9e74073ea2 : SUCCESS in 15m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/13158436e1824523b1fe554e7e91d48f : SUCCESS in 9m 18s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b607e8cadea84505a395c3523836c4aa : SUCCESS in 7m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/95d0e78d442c4e8e95276e2bc1fdb144 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a75894db0d34503b94295db2ce8e538 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e6863137cfb342b39656e2aa733bd9d9 : SUCCESS in 8m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/31cfb50184ee48899f5f24769e0953ff : SUCCESS in 14m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8da4b0b937f84a12837a99466531bf90 : SUCCESS in 29m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2123ee2fece84fc5be307544014e3618 : SUCCESS in 12m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e050efc12dcb44578b097e405a2d5fc1 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ca96d2697a3d4a5ba1185035e86f2091 : SUCCESS in 18m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/00259f157877402c91da692e0ed416c7 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7da33b6e8abb495b8eb78e401989ddaa : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/106a568c48e945c2992ce57d8edabd01 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e540fcfd4c4d48fe9fbcd514c01aa2f8 : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/af8a0e9b83624a4f8492ce35568810f2 : SUCCESS in 1h 07m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7ae8bc0480defd79a97d29fbe440b5fd383a1695","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:37:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78a3c9897393cf8757d2991bdda69f8c2d071b2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 18:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8f691c99e83446d83ced85b42d4478d : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1},{"id":"38f7f197725801418fb13238327ddbf608af0cb5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-18 18:43:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"05efe3bce012e3b5ff06b95948c68676690c3b6e","revisions":{"05efe3bce012e3b5ff06b95948c68676690c3b6e":{"kind":"REWORK","_number":1,"created":"2021-01-18 12:19:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/92/771192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/771192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/771192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/771192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/771192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/771192/1"}}},"commit":{"parents":[{"commit":"097e2110b0548890309c548b2117647965c24fe7","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/097e2110b0548890309c548b2117647965c24fe7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 12:19:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 12:19:28.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.6.1","message":"update constraint for oslo.messaging to new release 12.6.1\n\nmeta: version: 12.6.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I3062abf38d2581a66c363cc7a6292dc0bc16b4a2\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: If72068839df8fc39cb6f923f3e914f6db5134210\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05efe3bce012e3b5ff06b95948c68676690c3b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05efe3bce012e3b5ff06b95948c68676690c3b6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
