)]}'
{"id":"openstack%2Frequirements~771276","triplet_id":"openstack%2Frequirements~master~I0df71f4193d81adf2259794ec69c2670bff294c4","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0df71f4193d81adf2259794ec69c2670bff294c4","subject":"update constraint for ovsdbapp to new release 1.8.0","status":"MERGED","created":"2021-01-18 22:03:29.000000000","updated":"2021-01-19 06:17:36.000000000","submitted":"2021-01-19 06:13:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771276-new-release","meta_rev_id":"460642901e8f810b400dcb2da1c42d5a95905c45","_number":771276,"virtual_id_number":771276,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 06:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-18 22:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-19 04:29:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-19 04:29:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 22:07:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-18 22:31:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-19 02:30:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 04:29:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"42b69daeb4a766953eb4f5cff73190c653880252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-18 22:03:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ada373b2f73f50a6eedf608eb89913291e1ebb9","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-18 22:07:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a2136e07fb06013e6a85cfe68fe58c1cfa38495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:31:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f4bf279349174cfe925a412be85ec773 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":1},{"id":"76585e61f6829ad13727ddbe71a9f145fee70afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 02:30:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae155a880d14391aa978e28cc3f2b54 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e51465493d4350b93fd394a9468992 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19191255e21d49dfa6d7464736ae9800 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b9b161059842b5884ca2a2efe4fdd2 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30f829096f804e8f982f68218307cff3 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbf5ceb1c3464981aac940979e942c85 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c026d373b9e848128d5e141f66b2a441 : SUCCESS in 3m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ac568513e52f4f63bd92cbb73238f4bc : SUCCESS in 8m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e0c14fa546a44e93a4e87282df97d7c0 : SUCCESS in 12m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/27425f3cea3c4d568e70f26c8fe03491 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3d435646fa54fb189d963ae7d1df6af : SUCCESS in 4m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/37179402553948dd92f0f4b48e7d230d : SUCCESS in 8m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/34db5afbca0543bb94a92a6f576615de : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/72be7122147040568f5d0c7eb59b0ea2 : SUCCESS in 7m 22s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7651953cc65a4a59b5e3acfd93db7524 : SUCCESS in 14m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/53852b71884a4e1da4a2b89018f2a726 : SUCCESS in 28m 13s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/94ef68a4c1bc4aca98c6f415902630a0 : SUCCESS in 11m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c15bce1d2c54fef9b99cee8b3222bcf : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/00a5ba472e2a41a8af72d5795da1968f : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fe122c6e55234697be6281614906b3af : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f358bb04e7a647e6858d000e3d232228 : SUCCESS in 5m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ed61caaef90b4d6f8bdb7a8d4466c46d : SUCCESS in 10m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c74a56a7c1f94cdd98c216b157236948 : SUCCESS in 44m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b154143b48d944208d247bd3b8b28e5b : SUCCESS in 1h 24m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ac444edeb0813a4d0401f1e0263a2a50a7acdf56","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-19 04:29:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5708d84fb9f72916751e131473ce06433b8554cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 04:29:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13cf9b8b28abc553546ba30b15ff1578170e9eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:13:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3cbcd288c9849b39e08188f7aa7610c : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cc1c3e3e2c6454e88ebe1dc6c65b10e : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667a1a1c6e354920ae1057b8b31c912f : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70482cc3d5784209929af876582d5502 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ff1fcdae9ee464e963ed32b81913167 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0f098d93057a46ca81b7884f6fd107f8 : SUCCESS in 3m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e0e9877c18dc4e10a4b2f80585916b41 : SUCCESS in 8m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d9bf3fd272f649139bc9048e445b3415 : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c4fff41aa55a4f53b21b3366ab6d4d89 : SUCCESS in 9m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0ddabea21c574fb9b128b6fa99073c2e : SUCCESS in 4m 54s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0fca4055c0cf46f1a3a3153ac7df4e94 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6257f224991449bfb882bbb9865d63ff : SUCCESS in 6m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b974a66a0f90455aac2dae740e758aae : SUCCESS in 7m 03s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05352466103e4f0ebea34e79584edbd6 : SUCCESS in 12m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e86f73eaf3b14e3784931cc930040846 : SUCCESS in 27m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4eaf84a51973436eafdbd929e730e786 : SUCCESS in 15m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc5e6cf66e0d410fa32dfc563634524d : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7eb65af8e474c749c3de348b570bf91 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c13428c6b84245b08d7a69697edb63ee : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d073de2d50bd42e8ad400a641e548733 : SUCCESS in 6m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9310c3a11c6744428dd736346e4df93e : SUCCESS in 11m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e149cea0ae43fab9ef5a5e31cd3a0f : SUCCESS in 56m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/382fcb2394c04b90b64bbb50b5e65811 : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ac7fbf494446f3ae264b539b6678e13c31b86578","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:13:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"460642901e8f810b400dcb2da1c42d5a95905c45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbcd07b003643aca7d398a5d1641fe8 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00453ee8eb6a2ae7a6a12b62f99b7b7d7f14a66a","revisions":{"00453ee8eb6a2ae7a6a12b62f99b7b7d7f14a66a":{"kind":"REWORK","_number":1,"created":"2021-01-18 22:03:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/771276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/771276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/771276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/771276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/771276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/771276/1"}}},"commit":{"parents":[{"commit":"05efe3bce012e3b5ff06b95948c68676690c3b6e","subject":"update constraint for oslo.messaging to new release 12.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05efe3bce012e3b5ff06b95948c68676690c3b6e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 22:03:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 22:03:25.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.8.0","message":"update constraint for ovsdbapp to new release 1.8.0\n\nmeta: version: 1.8.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I20e8a80e03497ed69e1f64556ab28a5d0b45efc4\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I0df71f4193d81adf2259794ec69c2670bff294c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00453ee8eb6a2ae7a6a12b62f99b7b7d7f14a66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/00453ee8eb6a2ae7a6a12b62f99b7b7d7f14a66a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
