)]}'
{"id":"openstack%2Frequirements~771279","triplet_id":"openstack%2Frequirements~master~I57edd0b78b1b02893f33194a21c907f530468cfe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I57edd0b78b1b02893f33194a21c907f530468cfe","subject":"update constraint for python-freezerclient to new release 4.2.0","status":"MERGED","created":"2021-01-18 22:09:56.000000000","updated":"2021-01-19 06:14:07.000000000","submitted":"2021-01-19 06:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771279-new-release","meta_rev_id":"a1730c10bacfdccd6a897c973a099f49172293c5","_number":771279,"virtual_id_number":771279,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-19 06:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-01-18 22:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-01-19 04:29:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-01-19 04:29:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-18 22:25:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-01-18 22:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-19 02:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 04:29:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dab143df16900599fdce68f0a175604ed9111e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-18 22:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81aac63e207281de167ca8cd899f1c20c6fbce29","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-01-18 22:25:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ada226f20ada0b85c9e3166c687232bfd0b2b376","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:30:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0f3df314f6fa4a6792752c6ff96085df : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f04ed90607ba4dcac92f12d653edba728ea990d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 02:43:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1d9fab7a2946e5ba0f17e7e543e728 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8436bb5fd9d4807b1d970c3c399a279 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2155a4e1981e41298c261a4f432e66c0 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8bef0f879a5471fb4d2ced9a2918f07 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3053254320f54813a466b75ab6995385 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a700bc30dcc54936a87862bc608d06b6 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e04e47edf6c44517933c17fcf7175828 : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8b940c35ddaa4cdfb8d32e1daf17702e : SUCCESS in 9m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e086124668f14212b35951e26c66703f : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/31e928bc26964728b0ccd0811953ef2f : SUCCESS in 10m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5bdfd8c6c9d484b82c956e2f0859e74 : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/620def1f64484b47b868f249dcfa239b : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/894f2a14af634ca3ab6b9512240ca0f6 : SUCCESS in 9m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0ab39872b99848228a3b0b0b8496f50b : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0d8c47a8b3564d34b419323140f8be4b : SUCCESS in 13m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e18046922ddd4e83a5d1d90f4855e5ac : SUCCESS in 26m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c2b4679340d14cb19a525e9882b5e970 : SUCCESS in 13m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/36087c41e9934042baed0389eac2a333 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1103dfe8e8af4ee7b708acc81abe53fe : SUCCESS in 18m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b0caa09fab584a2389b0f6de89307bff : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5ee2d94485644b47a572733452010a43 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4384eb9be58d42a7a4ca9623b9751c18 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90259c8db8ca4d20a6716b5d542629de : SUCCESS in 1h 01m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/786412f834d64e30a929441ff56879ba : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1d9cca81685a902eef2886e59d165d3a8d7be007","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-19 04:29:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"560ce7c0d38bc5da0ea4329078e903734971c4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 04:30:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da6ac58d956306ad1a0560802644ebc411a0aedf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:14:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c5f18f92fec4e96b326cc81844df8d0 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1a10fc5c894e8592aadb948836ca08 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6228cc4e7489487285a7822560a71281 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59a242f974894865855101369d87e0bf : SUCCESS in 3m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae2bcbc90131433c8c960e3104ed1052 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/90f646a8e474492593f07e16f88ab058 : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1f0170f085134721be38d503082e3b1f : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1092f6436027431e969a9a731a84ffb3 : SUCCESS in 12m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9269773d39c64ef2ac56eb5b33a14f6e : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e8ccfb9042b04644ae926c73f31f0f50 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b66c8872b5064cbe9b4bd88eb1214335 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff988e0e0c00452eaa31729c7d09da51 : SUCCESS in 8m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/030b47ad26bf4b5fa1e49f68884c0772 : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/407597c91884456b87d2d0a2ef64ca78 : SUCCESS in 15m 01s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7dfda3aa6f624876ac2676ab658d304c : SUCCESS in 26m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/457c668b3a7c4373a47a40216c8443c4 : SUCCESS in 14m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a74735b0a4eb4a79b77740a2831d7f4f : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a45b18af1e52459db2d67a710f6aeb82 : SUCCESS in 20m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c35afdbcfd4c48fea2142820fd13470d : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80aee5cdda144227a2b95ada21343526 : SUCCESS in 6m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/502a472ea44d456bbe0e98a9fb6737b5 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59f291222d26453c90a8d72bfe17d75f : SUCCESS in 59m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b3ba5db8c07421cbc905176fe06fe1d : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a1730c10bacfdccd6a897c973a099f49172293c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:14:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b75de49b542cdc64693ea27dac784c7b77bfdc1","revisions":{"8b75de49b542cdc64693ea27dac784c7b77bfdc1":{"kind":"REWORK","_number":1,"created":"2021-01-18 22:09:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/771279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/771279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/771279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/771279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/771279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/771279/1"}}},"commit":{"parents":[{"commit":"05efe3bce012e3b5ff06b95948c68676690c3b6e","subject":"update constraint for oslo.messaging to new release 12.6.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05efe3bce012e3b5ff06b95948c68676690c3b6e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 22:09:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-18 22:09:53.000000000","tz":0},"subject":"update constraint for python-freezerclient to new release 4.2.0","message":"update constraint for python-freezerclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I50c5875efaa52ef1ee449a001d3cc6dcc23ecd84\nmeta: release:Code-Review+1: Carl caihui \u003ccai.hui@zte.com.cn\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: I57edd0b78b1b02893f33194a21c907f530468cfe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b75de49b542cdc64693ea27dac784c7b77bfdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b75de49b542cdc64693ea27dac784c7b77bfdc1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
