)]}'
{"id":"openstack%2Frequirements~771359","triplet_id":"openstack%2Frequirements~master~I97830222ff9b912e640bd6c4659bf9a643f59410","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I97830222ff9b912e640bd6c4659bf9a643f59410","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-19 06:20:26.000000000","updated":"2021-01-20 07:42:34.000000000","submitted":"2021-01-20 07:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771359-openstack/requirements/constraints/noclob","meta_rev_id":"a32edb0bdf1a73e0ff243bb32d3421f8da764565","_number":771359,"virtual_id_number":771359,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 07:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-19 16:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-19 16:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-19 06:30:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-19 08:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-19 16:57:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"736c18621b722a2d2d8f6683630e6a9d25a93cd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-19 06:20:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"061de6b9834ff0aa3bb382488ff7cc6654d5dd30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 06:30:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3a4dc995a1c84068b3d17a7e4e50c63d : FAILURE in 8m 21s","accounts_in_message":[],"_revision_number":1},{"id":"b79ba7e4cfaa371b728fcf047c69817741eb9b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 08:10:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a277437adee74151a68d57a33728cfb5 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9bb2f239034138962e3216409fe824 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b91c898f0a49f8bb158137bda27f37 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52485865f3824c999842c7410128c52f : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbfe995891894eafa6ef21f5212b4462 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2d239fa74f52465dab97a6e627eed834 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f99cc1c836d43b383fbf4709984c54c : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/19e5a2452db64ecaa954af3b63b00c4b : POST_FAILURE in 4m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a3368a7be49441e9a7980328bbed68a : FAILURE in 4m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/95ffc0437fe840489b1e633cfa5cda16 : SUCCESS in 8m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2cef6d49b6b442a1924299c76069669f : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1766ec60e34a46cead7ecfc1781a9a98 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1e949650135491c980e49eb110d61a2 : SUCCESS in 7m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1311f9bb23924a10b2633c1cbb1874ca : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/556d567f72b342ab894cbf4a38bc907a : FAILURE in 14m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6f9356529b7847d9a33cc01eed3f4910 : SUCCESS in 16m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4ee9db14c2674799ada2364ea3f11531 : FAILURE in 11m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/855e5968155543b5a9f7198a2043e538 : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5dd00857500443e83d552e63f558b0d : FAILURE in 18m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ae23b2e601e04355a5c91cbc0aa0c9ff : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6ccf52e9022a404ca7477219d90dda22 : SUCCESS in 5m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3e3d9327b9174cf4bb97204fbddbe94e : SUCCESS in 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09c93875b974473fb48643e46af74a09 : SUCCESS in 1h 12m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ad4c55eae644be0ab401a74a296e950 : FAILURE in 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"770995ac571f14da7aaca1f39bc0d816d51d8cff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-19 16:57:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7afba71f33b17fe646c0fcd898e158a6c9e6dbd0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-19 16:58:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"248ceb522cbf250d9a17fab47219ba873c389fa4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-19 16:58:12.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"ac79ff0e91c31e58a55a513b045c6ad3bae903be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 17:31:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/81aba8ad01924db3bed6f7ba6ca99ad1 : SUCCESS in 23m 05s","accounts_in_message":[],"_revision_number":2},{"id":"29251ab8e6f4b071cd5e0bf1a3acfbbbb70c9b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 03:59:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd3e3d7f5b04d4b9014d77876a08e34 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7fa4b7d07454698be569758de433786 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cad0f070a0047ae9a3526fd95860d61 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5776d77142694ced89d6713ad9567bfc : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6edfa15aa8e3436fb8d62f4e73ae9ece : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/489327de9c3f4bf282f357be89c60053 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5dde52ade2dd4efc8a4bf74de869c3e5 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9adf09ce44a4014af2444316b4025f0 : SUCCESS in 8m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6b4b930e2ce841178acbd23efb100fa0 : SUCCESS in 9m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dd5a6a5b1d3240d9b9dcfd515664ecb3 : SUCCESS in 12m 19s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bda2aac578434d1d83054ed5db8f4308 : SUCCESS in 5m 32s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/45b3858a23e340f3a3563914b495848f : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8801d38bf02f4d39bd04a72948e4e5be : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7716f3b7b9804026aeb4f676813c41b3 : SUCCESS in 6m 57s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ef96103fcf3c48098de4b38af7862143 : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3e2a4f4577dc4e27adc4c5bad9fae831 : SUCCESS in 26m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/464a56ed24534314b3110d270cb3da4c : SUCCESS in 24m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/301266b286e34709832787779717beeb : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f85af2f614c94f0c8804031e5525a24a : SUCCESS in 17m 49s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/debc7413afd84a2281b48e0f05db9b02 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4da17c8cf594d248bdfa9dba42f104b : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/9fb173976a8c4ecfa710c2621c9c193d : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd993b738fb543559355c75422c34855 : SUCCESS in 1h 02m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2abc5c365aa44724b7e5f5e8fab3c0b0 : SUCCESS in 1h 24m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7292ffff19aa71097b7ec2275032adb73fe1a556","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 04:00:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7e850e88166293811df1dc61aaa8877bad57107e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:37:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472a48faffd641bda85a3d1bbd2151bc : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4063929a29524feead736d81d0895058 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d08277dd04c04f5e8bc2049306230213 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f58167a41034e1e87d672030f18ce0e : SUCCESS in 3m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efa82ec2cee74d8bbcd5599bdbdfe54d : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d99d3108068b4496bc4c2cccb3ac39e6 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d7fdfe66a22b491dac762e921db68a6c : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8100766072c142c8b7eed3953a709895 : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f885faf677ff42c896f4bc3c8f18254a : SUCCESS in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8150bb53b285456b891245ea2ab158e3 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/041be2e6fe174b628fe1614b6ad3b129 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/790a47425b944e549a43763dead854a9 : SUCCESS in 7m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/46666bb054344666ba4f88134b9dd6b3 : SUCCESS in 15m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/589f55aeef7b484a85e16c714f25c339 : SUCCESS in 15m 10s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6d28bff9c7aa40a3a40d7ec37c68acc5 : SUCCESS in 25m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fe772f16550440298c612244eb3bb4e1 : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9e5bd3eb9ea84ebcbdb5091f72b65d9a : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4e5809c6ae82450a9fc90e8047195a64 : SUCCESS in 18m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d8bafe36de8f4dff9e118002350a8505 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/84bdd7e157cf4f80bd70a83d1b337447 : SUCCESS in 5m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8d29c1815a9e43c59637a50df49b9eff : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae73f7345346416b87ca38c95530af24 : SUCCESS in 52m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/61d17c84bd1a4bf884ad96fb1f728403 : SUCCESS in 1h 27m 26s","accounts_in_message":[],"_revision_number":2},{"id":"caed1ce773ba7753a8ff206df574d5fa311ab938","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:37:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a32edb0bdf1a73e0ff243bb32d3421f8da764565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 07:42:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4952f71c5b147ea98b468be4b83deee : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"961c951c30e3c9c1e2ce15faf5b464423858bbae","revisions":{"9b68ebd9c65883409dfbcd376f85c57e47b7ad12":{"kind":"REWORK","_number":1,"created":"2021-01-19 06:20:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/771359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/771359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/771359/1"}}},"commit":{"parents":[{"commit":"354c5a699cecf766d3313143d80b709953c00f3f","subject":"Merge \"update constraint for osc-lib to new release 2.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/354c5a699cecf766d3313143d80b709953c00f3f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-19 06:20:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-19 06:20:22.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I97830222ff9b912e640bd6c4659bf9a643f59410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b68ebd9c65883409dfbcd376f85c57e47b7ad12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9b68ebd9c65883409dfbcd376f85c57e47b7ad12"}]},"branch":"refs/heads/master"},"961c951c30e3c9c1e2ce15faf5b464423858bbae":{"kind":"REWORK","_number":2,"created":"2021-01-19 16:57:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/59/771359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/771359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/771359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/771359/2"}}},"commit":{"parents":[{"commit":"354c5a699cecf766d3313143d80b709953c00f3f","subject":"Merge \"update constraint for osc-lib to new release 2.3.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/354c5a699cecf766d3313143d80b709953c00f3f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-19 06:20:22.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-19 16:56:47.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I97830222ff9b912e640bd6c4659bf9a643f59410\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/961c951c30e3c9c1e2ce15faf5b464423858bbae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/961c951c30e3c9c1e2ce15faf5b464423858bbae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
