)]}'
{"id":"openstack%2Frequirements~771753","triplet_id":"openstack%2Frequirements~master~Idabd9f6a071799b6900b7892e9b5d25dd4fc3500","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Idabd9f6a071799b6900b7892e9b5d25dd4fc3500","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-21 06:42:02.000000000","updated":"2021-01-22 22:58:44.000000000","submitted":"2021-01-22 22:55:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771753-openstack/requirements/constraints/noclob","meta_rev_id":"000a1ece6fba53c41cc96ef4ae50de925d4cb793","_number":771753,"virtual_id_number":771753,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 22:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-22 04:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-22 04:09:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 06:57:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 08:02:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 19:10:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1a21ffaa15bcad784ed9cb3b77a080f5429aa656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-21 06:42:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d2bf6dfc43de3088f8a5d5ee13a42a92ff3f39f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 06:57:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/29c411af0e6446cc8d8663dbdda1438f : FAILURE in 7m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3e4a4fd237f07672633506e805cd5a079ba19e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 08:02:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca22eff8cee4fee8c7bf7b16728b41d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f61697361a64eec86e692270a4d9442 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3fa8cb3a52498c802c295f694daf2c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5588775f2f74cb9a77614fb1c485bb2 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d7ed9feea9e4e26b199ade43220ebbb : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/43c836643d6147a591e6ea0fade77caa : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e85526f88994d8cb6e74392f4b1ee75 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da0021d75ca34079b1033682c3d0140c : FAILURE in 5m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dd1182d282e743c7bb77ae50643bdd75 : FAILURE in 5m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/db0061d58e50480ab50301ec3af875d4 : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f0c21bb5fbce4c609f67160f820a40e1 : SUCCESS in 5m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/108a9ed416cf492faf42d9625af97bf0 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/15cfe014d145407184932455c1a661dc : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40830c59bb72470ead7700f887adb56c : SUCCESS in 7m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5851efc9f96a4bbd83f70c1d91582274 : FAILURE in 14m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b2b3b766e5544ce6b25b4a561a7a81cc : SUCCESS in 28m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/48e068b800b444119803ee7beab1e9ed : FAILURE in 15m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/beecd4dc849a4f8dab3e030df8368f1f : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6a59543b3aab4f9599ecd5a36e9eabf4 : FAILURE in 16m 58s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa4a3e77670e4bbaa8af544a26ebf271 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d46de8abc5aa4237ba9f87ea0c8643eb : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ee3ca1d61c284385b9b24fec801288b8 : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ade5ef9b20e340e5ad1c1afd39a53b4b : SUCCESS in 55m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/003aa02c7d04490a873443b83a16d344 : FAILURE in 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e038833716f4b87b5606e6770f7dc6443379905b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 19:10:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"073155905176fcdc7ffdb5a6f75ec279bb8c3fe9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 19:10:40.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"8ccb533d6064679d10e6add1ac385b201f892883","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 19:51:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"e9b64032560636fdf755947a6065ec28aedd7d53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:19:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d559a91cfc4daeb44cbb0c8fda6387 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6794feb6264df7a5db3460efc8d072 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73010511042f4af899189a1fd22db980 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b45247eb8454422aaf0039d95604c14f : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bb982bf3e824f3e969b65f1dab8ea71 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3c48fcb9fb2b487eb85d406588f694a3 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b41abb7e805424a9e327f44693b3322 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0c28e60daa9e45b2a8455fc3d7263fc4 : SUCCESS in 12m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe05f25b2a054c65b299f74f980e157d : SUCCESS in 13m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a7ca1c4285ab4536b90c82116c051bad : SUCCESS in 9m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4969a6c2e073477cac728a8b87f90fa5 : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/187ebd2a204c48a5a2aba89e279236e1 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0e02fa1e09c44d2bb285b7d21e0b683 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85777d302b094cf4ba77c562f2963c7e : SUCCESS in 8m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/36b7c699ba414744a820e407b4a1d714 : SUCCESS in 13m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0614ada7ea004662a77ac1a1927e11ee : SUCCESS in 30m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/448ef970c0a54c69a32d0384626b1188 : SUCCESS in 14m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8fad64a5d2a448339bb5457dc1a3777f : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/992ffb479e2345abb0b35c5db1d448e8 : SUCCESS in 17m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b595204896bc49c79857ab2fc2b41c8d : SUCCESS in 7m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/53fc4bf13d71427aba3bb6e616f3917d : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ddee9e39224f45a8a1a600403648f8bf : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bc04367fe684cf49c67662274ed45fc : SUCCESS in 54m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877f4ec063944caa9033c673dda923a3 : FAILURE in 14m 01s","accounts_in_message":[],"_revision_number":2},{"id":"db0b7b94f94130c683b84a4ea25a7043cf9516de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 02:03:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1a5e0a8c1f78c8420a8d6ed882d6595afda693d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:31:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"5ae92ff19e0408490513d74fb1211d5aadce143e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 03:51:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693d145e7b9845ba92d93f8796051ad7 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c497636044b145b285b575c82aaac9a8 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c55ac116d5e049b389914cbf75dd9a18 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdee3d6d83584d18b70bef0d748674a6 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a217738e1224b89a71d05c4575e23b8 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9d2ac04016c4674892679a4fc132951 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d994bee0e0074eb8a39ba1980469f549 : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6043a7a9d559400493ba812b19e0f3b2 : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9d80f0ca4da14c1696ed84b8540a52a9 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/862b07bd51304c0ca32131da29c3271c : SUCCESS in 10m 24s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/598a6ff4197c4280a77e532f3e468020 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4db9094b537e459f92fc255ebb0f9633 : SUCCESS in 8m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/579f177054e64a9eb289359bb885287f : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1c029a489bd64cfb93e4d6338fb13598 : SUCCESS in 8m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/396ffe53540643889f456df75e09882d : SUCCESS in 13m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8321719de67d467c893d86a2d75fa27a : SUCCESS in 26m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b5f25d859e384adbb7f623d53be66369 : SUCCESS in 16m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/49e75e614ef34dfbaf63b6175422e0b4 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6a1d73c4e1c492a8e0647e0c08deb51 : SUCCESS in 17m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e9ce3e6d6a3f48378655ddb64db7907e : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/516c07b8b9a64cf38688f5168426b7a6 : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/74d58832ed03481fa76ae1215690c881 : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c27b30970e248dc92e04b1722cf12c7 : SUCCESS in 47m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82d155e7bdfb4bcfa2d9611d40522fcc : SUCCESS in 1h 23m 20s","accounts_in_message":[],"_revision_number":3},{"id":"334ba72342c6ad2a587572b00cf36b1ed14d8342","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 04:09:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f36a6a6dddb2ab76b360d4e49a6019537f0fd7a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 04:10:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48909e9b2583da82085b8a3bc66baac5ddcedc19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 05:23:15.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fd2b133dc0441308e51e51a7eb39dbd : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11bfc92afc34a9bbe04956bcaebac10 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbf46e1968c94e44bd3741aff25e2038 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccfe6b84bae14ddf9a728ec8b2c5895b : SUCCESS in 5m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e933f3d796840f3a4bee0ee80ed4328 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71c9f37efafb43268fd460df63ef4190 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/446fe02654654db5a936e50237b3463d : SUCCESS in 9m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38966014bca946d69a3798e6d142d016 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/408b29190e724cf5a131425dd137c9d4 : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aeecdc981d184d46804b8b36733460c9 : SUCCESS in 4m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6afc0b5d6355493db7937538003fb9b1 : SUCCESS in 9m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d9ba0ff77b04e588cf739fd1078df74 : SUCCESS in 6m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/775ab45f8a48418f90b8507312a01a1a : SUCCESS in 7m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c0fc91d6782c4409a8dd5ab23a323066 : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c7910cd14da84f4ca4e3bbeb1d67bc37 : SUCCESS in 23m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/37be151176874fa19f0439705f17ca81 : SUCCESS in 10m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/336968cc58134b99bfd821a78043338b : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1da8c87f34a344bc90e4d73f747da9e0 : SUCCESS in 15m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0abec26298e24233859dc5bdf47b134f : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f01d877ddc754e128730764cfddd5055 : SUCCESS in 6m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5b8a8100053145149fded86535f80733 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/040ec75695c44ecc8bfacf1542427f01 : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/832ad0f687e742b391234127f0899262 : FAILURE in 1h 10m 19s","accounts_in_message":[],"_revision_number":3},{"id":"d2f6cf26dba3374d6495fdbbc404509519d474ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 05:26:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cc711ae7003e2d961e87f06794e1fc95adf85f0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 05:42:04.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"1ecaa41b16ca5a825bb6c9920b91e78c18730006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 06:56:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c86d6bb1bac40afaa7b6c3d0a2638ee : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca4ca1f68de4fde840a36bc2e8aebba : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffd735770eb04b709c638630a7e29864 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8cca22ddc1b45cd926ddf479ef3c870 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e47653c7714e69ade28ea5c119c712 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0187943c9204356b4521b0f312098ab : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0deafa2a640c465f9f953aa9308b4948 : SUCCESS in 3m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/419b0dc141384341a72f2e59eced1ed2 : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a1a1ec6c58344b38c5acdfe2eb04fac : SUCCESS in 8m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/976c20ae03f941809a0395ecb13e71f1 : SUCCESS in 10m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/83cf0a4054564270ab538a2712137156 : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0a60d7a0549544f48e17d319baaf2e49 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4561f9736a374bb791afe32eba4ddba6 : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af137949febe495fac1b47e47234aad7 : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e23ebe14d59341f1bece214c150e2106 : SUCCESS in 15m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0e0c0cd43e024be19c4e2b587b9fdf81 : SUCCESS in 19m 05s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6e27d69644645ffa2ff57a5f7341976 : SUCCESS in 14m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fe601e212063450286a925a94ac18881 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7aa6120a19884fad95d1c39f3d3cd618 : SUCCESS in 18m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0af89ac9138348bb9a3be8d506ce4336 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e26dd616dc748d2ab8a6668d116d2f7 : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6f131b1a94d24c09ba2d821ebdfdc423 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72a60aef2d384d7f8a857b3f28c38771 : FAILURE in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0047afa5154d4a61b387dc0ecf002f53 : FAILURE in 1h 27m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9be9b31d2e47d624f5dd46366d4a28ff11596cb9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 07:13:30.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b9384230c8a3496b34e2ebaa4832f9d879f2f598","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 07:46:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9898bf841084eb68d4493bc7798d456 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6fbbb12c4f50b450324269b9f44120546d7d242c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:42:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c60ed0456d43f5bb2e69b73742dae5 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5511b69f721243a8b4be3f511816e471 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/935d3aff201847a8bbc7a626903edd84 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e953ff6c577c433b973b680a04d9add6 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb238e944201453d984a66ea05131a52 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b2bebb3a875464db976b4cdf38d1436 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3cc6c289d5247a99f63016f4620627b : SUCCESS in 3m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1acab7e3d52f489ba64dde6fb3d24696 : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf251390ba7043e4b58eff1cfdebcd46 : SUCCESS in 9m 30s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2639386532c8470e8f37cc0f30560153 : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2ec8a4776124f1dac08db59f2aff3fa : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/20db4a7b5d5a497dbe1f932a1beff987 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d092d49d67cf4b7a9436cbc51a60ffc0 : SUCCESS in 7m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22240f80fbe74f21a6288df43ed5542b : SUCCESS in 7m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/33edfa5caa6745d49ba05aef7006f12f : SUCCESS in 12m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/020fed8bc6224161bea447e5a618a1fe : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f57266769e13485088c14f3a89fa407e : SUCCESS in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4bd247dbf4bd4fbf8ea0a3b6397b8939 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82266fae8c014a139f8b47e1f8f04e81 : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4003b733371a46ae98b515878ad29460 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3de223c6f3354fbca80fa0c5bf9711eb : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1d42b14aec7046d7b0b21a7de2a11bec : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5835b1223349473cbdf97d4af8b21053 : SUCCESS in 55m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8275093e19c14038943071d9b7df6a4b : FAILURE in 2h 06m 49s","accounts_in_message":[],"_revision_number":3},{"id":"21568fe9481a5e0bc561eec4560a99deec34783d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 16:02:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8ba8d21d1fb9125a2583fa9b322f379611ae24f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:00:38.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"6c9e0bad5a18366093e5a981eb431b481d712944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:26:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4508a37e0d5140a98b08895b9b6d3e33 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e1287c4d2d4d5fb10999fb9ce87f9e : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb504217fdac4ec2afee8b4520261301 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b99e1018642448bb25a9f07eb309e33 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dffb7c13acc4e46a812a80c7906df8d : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1eded942b11d410382ce298224a451c6 : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e2596f308a7d4974973d6991a34c4a2b : SUCCESS in 5m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3ed8853214d84dd284d8008b2deb1d00 : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/141e739492474a05b7071058909b98c3 : SUCCESS in 10m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ae67381d118844828a6cc5ba10cea654 : SUCCESS in 8m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8f3f7ad845bf46cd92d3b170d9a650d0 : SUCCESS in 5m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ccdf47a4851b40bcb787c7a371f9ab1c : SUCCESS in 9m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fbd750a6b00b4cefad6502ab1ac3c407 : SUCCESS in 7m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/687df393c5ec4e608a066879a8ab2146 : SUCCESS in 7m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/95bfb31fb3fa429993b8ac53067a169c : SUCCESS in 15m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1792c94872c34540b89a44806c001091 : SUCCESS in 20m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b44a0ca9a4cd448db44ca14189bd6869 : SUCCESS in 14m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/73a1c5a606714e8c887b93d9cc29210a : SUCCESS in 5m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e136f881eaa540d78dff1608ed2d047e : SUCCESS in 17m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/fccb1afcb8834f4d843e874e10531442 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/efae5ab9fba84e4bbb258e6a20eef44f : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0987e52e33db47908d0949040a27a086 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/958e5b02efb14e8eb94b63fc6d506dfc : SUCCESS in 49m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65dfd3762c694d8a981d43317a631e56 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":3},{"id":"e25e625da049b28486bb12dbf28c60541ea971fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 17:27:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"055ff6f83a37b9cabd94942bd195ec6e11940995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 22:55:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9423af918af418abcd7dae598e30d2b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63f23a04f8294b79b326ccc61fc442fc : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e54e7df19bc45cd9edda12903c85d2b : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8589781b7ff49369434358017a069ce : SUCCESS in 7m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e24fd0644820414a85c2ff5455926148 : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/affdcfdc66f04e5e9252b7f893e88ce7 : SUCCESS in 5m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c276ed834cd74bde97a45acf1f3aee08 : SUCCESS in 8m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/74f37b89382d4c69ab41403041519dba : SUCCESS in 12m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a62a3fe7427d4d039c72c6fb21eece70 : SUCCESS in 11m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/54fa39fc64ba4a4b85127c2cad9d0e62 : SUCCESS in 7m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0354e504141a46189ed47014c0983c4e : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb61815299ad43c79fab912cab6f2622 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ebcf0cfc733e48da9ad599326633df14 : SUCCESS in 7m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/931289e38dbb492da4d3931cc12d50d5 : SUCCESS in 43m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f0b5363c92f941c0bdb0e26223786f59 : SUCCESS in 28m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/afe57c04e90245c8ace24898b5e42079 : SUCCESS in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/845a9a0862714d29868154cd86ebe386 : SUCCESS in 5m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b57c32f0c84d4164adab840de02dbf2a : SUCCESS in 17m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/86fcce229e9f4ba2ad44fdbb9d0508fe : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2b28dfeb5c754f0395f0c282a9bc7dcf : SUCCESS in 8m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/64c03aabed704c61927e870ee3fc22b6 : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afb3fdef9cec4039b767eeabfe1e367f : SUCCESS in 1h 16m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d21c2d7c3c754347aa4109bd9d415326 : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9434486076df2c69f9b34f25df5f8ec7ef6378a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 22:55:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"000a1ece6fba53c41cc96ef4ae50de925d4cb793","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 22:58:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d953777b3304827ac76bb7682fe03ec : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b8a4c4f6e1454c9f26eab68fe70b9e5e01d8650f","revisions":{"816b928e814cd382a2a6264b83ca353e23b62486":{"kind":"REWORK","_number":1,"created":"2021-01-21 06:42:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/771753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/771753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/771753/1"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 06:42:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 06:42:00.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idabd9f6a071799b6900b7892e9b5d25dd4fc3500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/816b928e814cd382a2a6264b83ca353e23b62486"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/816b928e814cd382a2a6264b83ca353e23b62486"}]},"branch":"refs/heads/master"},"3127dc3259f0a73dcf6ee4e0ca03f2b36ab6999f":{"kind":"REWORK","_number":2,"created":"2021-01-21 19:10:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/771753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/771753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/771753/2"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 06:42:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-21 19:10:22.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idabd9f6a071799b6900b7892e9b5d25dd4fc3500\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3127dc3259f0a73dcf6ee4e0ca03f2b36ab6999f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3127dc3259f0a73dcf6ee4e0ca03f2b36ab6999f"}]},"branch":"refs/heads/master"},"b8a4c4f6e1454c9f26eab68fe70b9e5e01d8650f":{"kind":"REWORK","_number":3,"created":"2021-01-22 02:03:15.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/53/771753/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/771753/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/771753/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/771753/3"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 06:42:00.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-22 02:03:03.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idabd9f6a071799b6900b7892e9b5d25dd4fc3500\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8a4c4f6e1454c9f26eab68fe70b9e5e01d8650f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8a4c4f6e1454c9f26eab68fe70b9e5e01d8650f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
