)]}'
{"id":"openstack%2Frequirements~771771","triplet_id":"openstack%2Frequirements~master~I78c3a87c06b9eaf5df4c93031a53f19f7c84f95a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I78c3a87c06b9eaf5df4c93031a53f19f7c84f95a","subject":"update constraint for openstacksdk to new release 0.53.0","status":"MERGED","created":"2021-01-21 09:22:22.000000000","updated":"2021-01-22 15:40:02.000000000","submitted":"2021-01-22 15:22:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771771-new-release","meta_rev_id":"7d598d78f45a60ef15264f681ab63b07a5f60b57","_number":771771,"virtual_id_number":771771,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 15:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-01-21 15:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-01-21 17:52:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-21 09:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-01-21 17:52:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 09:29:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-21 09:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 11:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 15:25:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-01-21 17:52:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"df8fa2ac87717bdc5b45ba85fb9675580972536c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-21 09:22:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d40b2100f1db60a163f219ae0fb05f2f69d84ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 09:29:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"12fcbf72d5bcbccddd9b67d290a469eec293bb80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:45:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/edf2e5bcd93a4afa887d24d4cb160232 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":1},{"id":"15b410f4559a687bacb259e57e102d19606cf268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1a70bbd90c4827ac829cf7c6fc5b7f : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a88d063b461e433aa324bd19c4429121 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c8ce33f0c5a4f56ba546e21ccc669a7 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/709e48314c55415ba4041ed7f4d947af : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9a1fd11764a4642af36974426cc92ad : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/079aa47def2041fbaf8e547189420a34 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/442f060d2e0c4d5696053c9d22cfbea9 : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/485b78a1e2d049e09f379cfec88cbe90 : SUCCESS in 11m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5bbf34fe3d6043b7afef3c662972a05e : SUCCESS in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/94cd4eb80a7346198867a3b05cb0a9b3 : SUCCESS in 9m 57s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b625b3f90537418082cf1cc09210b086 : SUCCESS in 4m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1b8cf7a8beab4917911e55a41eb07089 : SUCCESS in 9m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/085ff9c10ae147d19f6a5a482ded1363 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50be0a26169747b1b1d108509a354d77 : SUCCESS in 8m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e3317bc390ec4303bbb7d5bbcff4a2e5 : SUCCESS in 15m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cd782ad6227d4180a2627105fbb7a685 : SUCCESS in 24m 48s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7199d7c725a9435fa6fe154c7047eb68 : SUCCESS in 14m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1cbf0aac2ba846e0aeac95451dcbc960 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c2128c5754f44709bed35fae52ffbacd : SUCCESS in 21m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4aaa85d6574449b5936834bc405f7338 : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e4b756db14774a91b30ca48afba9bb47 : SUCCESS in 6m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7dc6de7ae7fd491cafe918da87246f6e : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b840acc41ff4d59b03fc274fba7bd13 : SUCCESS in 56m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8c03b8b3ba949d0995847604e7057b2 : SUCCESS in 1h 25m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b11bce31dd5296bfc1db431d476f22a62449e578","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-21 15:25:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"296c2caf5e41a0e1c01e5cc77c5431a4fa532627","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 17:52:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bfe4801072d588c982840d4fdf9e7040d94ab944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:52:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b39529c6ec2da30820fbc4c40603e0584699ccd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 21:37:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1897c156e24417d9eb480cf2912a079 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b318594bd1434c83bc1917257ba5fb6f : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e1bc38d6ba544c6b700c69cc3f58bd8 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78fd5c24a3b944cb9abfcd656aa82b35 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae8282a990114ba1bb4e46f4d6f7159c : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0e8d9e3c7a7402a847ff25b400f9f28 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/af33a737444443a0befdb4f3f437f305 : SUCCESS in 12m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0f12854386c148478820620ea3bedd8f : SUCCESS in 9m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5211adb2bec44327b41dc2b8c7da340b : SUCCESS in 13m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c3e6956c180f48dda362286ee528f919 : SUCCESS in 6m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/464af12f3eb24803abd5e67818048d73 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d08c02b66f964b588d206ee5c53fc788 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/90b1b6025b934212813d0c3f0e708036 : SUCCESS in 6m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e70ec0d87c5f4a22992b4131b9ab7f89 : TIMED_OUT in 1h 01m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bbe3f00d5d4145a180bd905ffb6001df : SUCCESS in 27m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b14f98fc95554aa68bdc8d0c40ee0949 : SUCCESS in 15m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c147868fc0246e9b22a6cfd7cad8b76 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a3c106a82a54c70be1e80ce6b56ec4f : SUCCESS in 14m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a879ede18c6f42cb8be74d69ea8f4308 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36f28bef0a4840c6a754bd2880afccd0 : SUCCESS in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/76c13ad568fe41a69eac004f89a93fc6 : SUCCESS in 12m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1eafe28f55704b6dbebf5b529aceb9b3 : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cec0279eddf4851a7b398ae5e4baaf3 : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":1},{"id":"4be2ad8b35801ec9c4ee2a3a3c57e26d90157260","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 21:43:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"07e254f0ee98bc1b228a95f11dcf1a4af1cd485c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 22:29:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"1a7cb09853f03823fec616015f25a7602da87571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 02:53:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2320a3185fac43949bc8da715b3f111a : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7c58c3b94c46a8afc8c74509a55860 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aedf5ae7fe845e9a06c809f17a7883e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c441d9f29bf4394adf82f2bb339e27d : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dce0aa965af84aa289710b6ad9b08578 : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8280b7ba9d2b4ed5b5fb9fc5c456f91d : SUCCESS in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2068c6f41e404c659c1ded2ca58be9e5 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0903fa4bbc7744adb435ab769f5ae161 : SUCCESS in 9m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/74546d61ed2d42bd909fce90dc577f3e : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1beca109be284358ae0b00374ef51bef : SUCCESS in 11m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/710148acc1a442bcb4962c0ba2ac327d : SUCCESS in 7m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/98124526f417418a8f4532f2b9bf73f9 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6c03230881df4e1992cf55cdf6350524 : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf05cacec6a845ee9e5b324fa6518403 : SUCCESS in 6m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/48bfb319bd604c659823e2666ae856b0 : SUCCESS in 16m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0ecc18e244cc411397f47079a345c255 : SUCCESS in 24m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/57893d83615d483fa0dc93e281170283 : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9fe7b6ec0c9c4d97b4b286c058ef17bf : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6af5889782284a0f953cacd97733e20f : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87105213b01c42f8abb765e1183b5f03 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f11feef138847b1b1b70f46d29d3d8c : SUCCESS in 6m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ac13d6db3dbc4932a956d35452843b0f : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a591161f9224d69ad49ccbfcfd6bcdc : SUCCESS in 59m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65f1b987247a4f50a4c17901d0fe30e9 : FAILURE in 1h 09m 10s","accounts_in_message":[],"_revision_number":1},{"id":"a38d5e518bd2f05e80354233c0b4bc0c50d5aa84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-01-22 07:58:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"430b11965990274fe2df9257b204a73a36d7cf35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 08:25:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2492e2d18e4247739d512341214a36f9 : SUCCESS in 22m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c05727e264b9ef52b9512ceb480ab631630b3cee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:15:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04855eddfee142508e09ef2bee9b56d4 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645053797d0b4930be262375d5d554d3 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49fc9fb656ed460b880b611d37401b9f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9544bb0cae6e437faaf10f3e47929660 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54f6a9d349c847f7ac57c2dcc6c0cf11 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a23b64f3ce6f4ba891fafc5a496af1a1 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdf82cf3a9c7406d85ccb889e0b2ad23 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/fdfcc632de664e9596c973390d0d060f : SUCCESS in 8m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d6441de353e34a0a9d784e76b319c239 : SUCCESS in 8m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ce6a9749accf4bb8a2fdc94fe5549009 : SUCCESS in 8m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/af9489496ab841b3b4a12163e59bbae6 : SUCCESS in 4m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8d92b396539c4e70a0393ab0d3d1002c : SUCCESS in 9m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/58252a9149414f53acff6eccf3f7df39 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8a4e27cd42964512bff4bf2b08022a62 : SUCCESS in 6m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/883c0e7dd7474f0484c849810686388e : SUCCESS in 12m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b5deac53019148a3a1194321132afae3 : SUCCESS in 22m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1b499fa2ddc5478b91e9c09b6ed70ee1 : SUCCESS in 12m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e73ba848b3234ccbabf048295075a893 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e34a30e4069040dcb05f32cb73b489a4 : SUCCESS in 16m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/22fe67777b5949638e9fb9807536018d : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6e02f1542aae4a8db98672d1abd46d96 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a4a0ad80f3c249b79ad73ef8c4a9f5ef : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8baa90d88cf45ba80d0bbf2a2779873 : SUCCESS in 53m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f86681b020b4c2485c293d5337d12b0 : SUCCESS in 1h 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"61cb7b6f47a0fc75663248267669051fc63ce00b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 10:15:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"408906daa2c1761dca25b2e470ce1ef5a35993f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:22:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fdfe2f1941742a78f9a480c82fa45ab : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ae21b4deb8457c9839fe3ade699795 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03af074d2f3b41f1bbb79ca90d22dec5 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19fc4e8ee0f4c5096053eaac249735a : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0e6d2d2d143e437e88e2f085583a89e8 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08def6bd5a6945fc8d5c3b5529ba166a : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9cb4df83d91d4490b95cad50b8636c35 : SUCCESS in 11m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6be752733ded440c8166c81cd1b69070 : SUCCESS in 10m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ba5dea1fff1143cbb33fb0cf9ce2cd4e : SUCCESS in 10m 12s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/610b4f2be6424fff883f9c768d1e6323 : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d10e6eceb65e4926adacffbb0cca586a : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1d10d45d5e574a47a3f403e3ef51db6a : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6d55bd1ce7ae40098e9f9eaa4a91894e : SUCCESS in 8m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/05ad8db302fc48168b242c06165c58ca : SUCCESS in 17m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4ffd17d6636143cabf58422b6255c201 : SUCCESS in 26m 22s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5a6d07cb57d942228aac45e0db7a4f47 : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a41747af836b45d78955618a6de9e49f : SUCCESS in 4m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99161335683d4c04879ac850c0a7146f : SUCCESS in 17m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc85a355ccea4ca2858b3d6d522fc9e9 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32e618ae22d640b1acd25b2a11c297c6 : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3ee0d62076dd42b99f908f15f47321d0 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be3a7a65a21444c4b37c1a97f00f8913 : SUCCESS in 53m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42226e597c6047f78d438a2afdd65c53 : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b5c3fb9da48b3ff05378b5e3c7df878686a17562","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:22:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d598d78f45a60ef15264f681ab63b07a5f60b57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be5bab3207104685b702f00c0fc5c7c8 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"738c2ce5c65e72e5b824c2dfd26137c9ba5c9e31","revisions":{"738c2ce5c65e72e5b824c2dfd26137c9ba5c9e31":{"kind":"REWORK","_number":1,"created":"2021-01-21 09:22:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/771771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/771771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/771771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/771771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/771771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/771771/1"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:22:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:22:20.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.53.0","message":"update constraint for openstacksdk to new release 0.53.0\n\nmeta: version: 0.53.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: Idb4a7f7c88d712354396a016bc2c54508c5a4a5f\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I78c3a87c06b9eaf5df4c93031a53f19f7c84f95a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/738c2ce5c65e72e5b824c2dfd26137c9ba5c9e31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/738c2ce5c65e72e5b824c2dfd26137c9ba5c9e31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
