)]}'
{"id":"openstack%2Frequirements~771772","triplet_id":"openstack%2Frequirements~master~Ie5ef2b2a7a8386897690ab9007955be09710f8c3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie5ef2b2a7a8386897690ab9007955be09710f8c3","subject":"update constraint for python-ironicclient to new release 4.5.0","status":"MERGED","created":"2021-01-21 09:24:34.000000000","updated":"2021-01-21 22:50:17.000000000","submitted":"2021-01-21 22:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771772-new-release","meta_rev_id":"b2ec17c6952c2aba50dc4979b965d2ff9400caf4","_number":771772,"virtual_id_number":771772,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-21 22:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-21 17:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-21 10:13:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-21 17:52:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 09:46:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 10:13:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-21 11:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 17:52:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2597e3e9fa347347ad4d2c8acf005ae1b48b4dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-21 09:24:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9d3190f99f41d2c56ac01ae3d350dafbc1aac72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:46:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1da1ff58cca54c7a8087570f09b47f29 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5be2e72a31f308c8bcfec1cd1da8ede18adb7301","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 10:13:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4b9bb3ddaf2193fdd8cf3cad0476bd5d6fe2cb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:50:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee2eb07ed77407bacb923ad33bd71be : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c998f1fac1458d901cbc1d9e974c38 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd3f3451bd694732b961e774ba725636 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66c93811f62b43bd8a4d6695f20f4da4 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08f602da3bad4689ac2e670457f7c18b : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/afd110f30b464c539a93520df37d9ae0 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d2dd431f06ce471195e11326eeb92be6 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3cbdf2f5d621472d9b2cb74117ed2e02 : SUCCESS in 12m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b08f56c247a6454e92bd8765d0ddd75a : SUCCESS in 9m 43s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e06b0054847347efa00c2e23afc16371 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e7d23c2231ef408f945d765e96e6712d : SUCCESS in 6m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/13e45bb8e1a74a0e992821375d3a0d2f : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/367e6bfaba234a5b937090372c421b28 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84dc899da1024fb9a8f56839981fc7fc : SUCCESS in 8m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3059145cd63e4ae99d387dfc3142be9b : SUCCESS in 12m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3f243c309b45448dbaf7532ba3a49f03 : SUCCESS in 27m 01s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b9f63741743746cb8f9a7b935b60dcb8 : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b91b6782f4cb485e89a6b339d715b090 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f6c4b6fe1219420e95840f463e9323f3 : SUCCESS in 20m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/91d6d2fdcf61421a82bc389ff7e7e661 : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33c909d532c142c89033f5411389d7f7 : SUCCESS in 5m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/144c1c5ffa3f45d4b3bee1e979a8c6a1 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51a3707e2a7d4564bcea2de009d46930 : SUCCESS in 48m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9c2aac59f7141078b96e980d4c5863f : SUCCESS in 1h 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e7cce30b26519224eba90e59f76b001992701824","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 17:52:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db6b6fb0370741c4cd002ce8a6f26b95f6dee055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 17:53:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4386c926588fc56dd0f9f8e68bccaeac575d6ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 22:46:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c6e2b5caae45e98758b2523850750d : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6040c3e51942daaf4893f34157d8a3 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/056ed014d0f245edb09ae25bd856a432 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3ecc11ae94a4d34af4af8df71c1b483 : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45e22c8bf0554c12a858942aa68e94e9 : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ceb313fa068d4249a51c3e2c5162acbf : SUCCESS in 4m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6f27716083f64bb8865fa36cbd716873 : SUCCESS in 9m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/820703f49612427c871b1bf0f2f9dfe0 : SUCCESS in 12m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0da326b4599c4cf4acdb4b5110c8a611 : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0bdacca597b34db186313ec69a9451b1 : SUCCESS in 5m 11s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cce0dfabfaa44d79b2c4eb074340d383 : SUCCESS in 9m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ba47db7e0e314317a413217b0ca407d8 : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c65fdfdc6d5c4ddc8fd90cbb60da58d5 : SUCCESS in 7m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cc355fcedf8b4fe9bfc6e42191ec5ae1 : SUCCESS in 13m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02f8656025ae4c46833f856e77025ada : SUCCESS in 25m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b3b0dbcea78144f8a7fced72202ce19c : SUCCESS in 11m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f6c711f9da74e4e96375384023f9cb9 : SUCCESS in 3m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a0d8dea4ea041bab012b465f9095541 : SUCCESS in 22m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c0192838ec8d46338ebefa7f414314db : SUCCESS in 8m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f0990638b8942bcaa6166ee00e5a4c5 : SUCCESS in 6m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1631a5187eeb493ebe0c28972f8ed849 : SUCCESS in 11m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cb042eecabd4d12b02ec021a76b6706 : SUCCESS in 53m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e003e83039dd42b69b3ba51b14f0a5ab : SUCCESS in 1h 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8564020c23abfeaf07d7800bfe1241fd52696884","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 22:46:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b2ec17c6952c2aba50dc4979b965d2ff9400caf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 22:50:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37538ce1002e447b86c6d96495d1df04 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"250cfab0be0235a24711305db28e5cc990e92422","revisions":{"250cfab0be0235a24711305db28e5cc990e92422":{"kind":"REWORK","_number":1,"created":"2021-01-21 09:24:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/771772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/771772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/771772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/771772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/771772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/771772/1"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:24:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:24:32.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.5.0","message":"update constraint for python-ironicclient to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iad2557888fe4d548ad32b2fa218104bc8ed42a77\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie5ef2b2a7a8386897690ab9007955be09710f8c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/250cfab0be0235a24711305db28e5cc990e92422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/250cfab0be0235a24711305db28e5cc990e92422"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
