)]}'
{"id":"openstack%2Frequirements~771774","triplet_id":"openstack%2Frequirements~master~I05046492f29774031012bafabd2733def97f6dbe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I05046492f29774031012bafabd2733def97f6dbe","subject":"update constraint for ironic-lib to new release 4.5.0","status":"MERGED","created":"2021-01-21 09:27:22.000000000","updated":"2021-01-22 15:31:36.000000000","submitted":"2021-01-22 15:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771774-new-release","meta_rev_id":"7186462190750bb140716579c995a72352e4f078","_number":771774,"virtual_id_number":771774,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 15:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-22 04:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-21 10:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-22 04:14:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 09:50:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-21 10:13:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-21 11:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-21 19:13:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-01-22 04:14:08.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9c23846567fa4d07f53f3bc7d99771ed3f4006cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-21 09:27:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f39f1490df8732059803d115a68eca5e5beee39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 09:50:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a08e9c5855a4248a36454d295da77d5 : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"9e2e78d685d8cdd24ef3ea7333ced3fca29a67b7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 10:13:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c0282478493d59ea5de89092a23b2f15b971652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 11:43:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a71395044204a4ea9b122d4e1449907 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe53f33a23d94385bae413fb0c7fbf31 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50886a7e61454d62857960fc719a4c4f : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6040babf3d8347d4904949742821d257 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed4c23328d214bb1818acea513985528 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6ad6c4664204be7ac5499d6cb97eec4 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f91a6e1f20af4e35bdf322e1ca5e80ee : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9167e49af73141d9839f13a1d01f9bef : SUCCESS in 8m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6418c755120b4138baf835f7717fe306 : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b57e988ca944b028cd20f95a758b355 : SUCCESS in 9m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a2fab99a61ea4618a7e628364c42de6b : SUCCESS in 5m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c4582412d45d4006a47db97ae2d0c32c : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a7419d15eaf404383a4c2588036a767 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3e838df5917d4ff280c52909d97180a2 : SUCCESS in 8m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/366bf30ce04541f3b94be237fcd288ed : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b64db997bfd44d7a9a3df5a5e9320abc : SUCCESS in 19m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a3998fa5e57480f8f23590b6bdc8051 : SUCCESS in 14m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e4c6dba9d1d45449ea80015a26e152d : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/10b652f3453a47dd8eb7191918e1905b : SUCCESS in 18m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/366cc9da72c14a5d8e4dee33494a5d25 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2171251c05e0450fa8fe85cec4c416b3 : SUCCESS in 5m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/720dbc18d049422782a09a7efc928d43 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5b27bbe17eb4f43a2bb88b1627c322d : SUCCESS in 53m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8916ced2e05457ebc0879a0f5e30452 : FAILURE in 1h 08m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ca1244ee06949ebe194fe0e28ae3c24d958aa04c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-21 19:13:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d3a8adeffd97d6d782e25c1a140e8dc24f9e623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 19:58:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ee618d19a4c20f60a34fe931e122ce916b673bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef9f5c465ac49efbc6047f06f3e15de : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b22027e2f9e4c48a0f597f3a3e44cf2 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47ac4617bbf648b580f501f7ff15df6b : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d908484e6014a569772c6c19e1709b2 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97c441a76a6246e5b69544224e7c3ab6 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a76d080d5ec94821906b3c2004106d59 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/757dc83e1f8f4630997f254ee07ad732 : SUCCESS in 4m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f8102cf3307f4e8baab5555df93c10ed : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e7f104fca10e43b58e682a3d680577ff : SUCCESS in 11m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fd1ce98d0c364478a2579d49ff8785da : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/929b32f138b74230bbf6e9600e781db2 : SUCCESS in 5m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/acf9a210d3744928b281432700b8dedf : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7604542ae2244d0cb003e7503c11fff5 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c798e1285a24d0f8a0740ca7a67b397 : SUCCESS in 8m 06s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/39150811b78b4a68abf2552847f37de7 : SUCCESS in 20m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb2fed50cb1f4c3486073012251b0405 : SUCCESS in 27m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2621bffd918640f2a1f9f2b334ca436b : SUCCESS in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/643026d9c707436581dc5e75585dba3f : SUCCESS in 3m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e8d53c8d22a2464b8c4021f821da9b0f : SUCCESS in 17m 44s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eda9a949d39d48709be192ed14f5bc21 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d1ea8f6d66804a01bdbda4896c87c214 : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cc476c5206ab4620a91af554ef211395 : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fed9b1acc4934da3924224fb6f5f434c : SUCCESS in 1h 10m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a43e52430614626ab1137cfa2b2de04 : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c594addca22a53013562c56005cdcfcefabd392f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 04:14:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1363f5f99f391cc1ffa41804d4ce7b47f8fce34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 04:14:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"933fec6da7611d1761cf88c468860b714a804eee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 06:57:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e2aaa904aa4d66a5d4227d90615d15 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c957b6b6f44121adb1dc52ffe32645 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9536e50227a04f1f93942a721bb7fff2 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/830f34ed681246b2b36895e0d17e41fe : SUCCESS in 5m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9f5f42603834b12a070873ab512d817 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b6a34fd3f41d48ea980f93f75c38018e : SUCCESS in 3m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/acee2724d26f4a3095753cc1b235698f : SUCCESS in 8m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c99c4d6608e44e499363da890aad355 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/37eef6f708454ece93433fed71bda314 : SUCCESS in 12m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/77a94529a6474e2eb638b8dee917be3e : SUCCESS in 4m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a6c33aaed4b541a08783f7c4d864852f : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c7e5258a92624b6a9cca53544a4a2e49 : SUCCESS in 6m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/020b7f622a484d9289217d57713368fb : SUCCESS in 6m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19305af5f83d4d6385e45133d5cacc81 : SUCCESS in 14m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99faaf12664b447993828d5c3721108f : SUCCESS in 36m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4667443a5ecd4d92bf348f8cd4505e06 : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ebfafe8a9e0d46a3b7d1414ea25b1e77 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/64b24a81749b4cb6811d651385afcf35 : SUCCESS in 14m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/794e0cd0a2fd4f5081ac0af953a87008 : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2eca008b3142482298c49326a454d54d : SUCCESS in 5m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd7ebce13fd24a9da4a3d46f74601b2a : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1fce8599562f48b683bb670d5cd7f046 : SUCCESS in 1h 01m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/469819f140254ef4b14d86b09b8400b3 : FAILURE in 1h 28m 07s","accounts_in_message":[],"_revision_number":1},{"id":"437c4f68a929cc5f9d4d98c36dd8e401bac0ce20","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 07:13:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c341014528d599091fd3b79baab44e36908830cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 07:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/74b243e6bb1e40de898eb8dcb163485f : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"54aef4f1f5df419994d1af60630f38bb28be5adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:05:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da904183577c49df91f3f968056fc3d2 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/229e6c82ce994aa6bc28174c93ab17b2 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db24ab8a99c945cf844ca50c184a32f1 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bc640fd03b54e7fbe0b2cd6ce2dcda8 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcb2f7a9e22048d588f9bd967476e0ff : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/87c5ba4248294f3abc3ebbe2d4e1e8d0 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ace2e0b3455c4585896b45c6ee197aa7 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efa63f43a27a43f78536c42042e25863 : SUCCESS in 8m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b144258c38f446dadd3f32b6af23e38 : SUCCESS in 8m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/514ab54d6a304a1fbbf9748835146ce2 : SUCCESS in 9m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d6aa8ecd6c74ee5b161e81f48c4c02e : SUCCESS in 6m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/68ea2f9c7efb4e059662051b3165b11e : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/be110149f59447e58b26fb03a865bc22 : SUCCESS in 6m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23e1d1e9df6045ef95516b8bc0445c17 : SUCCESS in 9m 04s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/da4396465c834f30a78f09c0e8f77c80 : SUCCESS in 13m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0db9f50870b94b1b8d75d9f7d2de6557 : SUCCESS in 32m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f8eceeb6f8bd41658f8508799e5b8394 : SUCCESS in 14m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/683cdbcf65e54e3cad8a77bb90d38b32 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a7369710ea0c43f680f90dbe1b79e4ef : SUCCESS in 24m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/424e87c8b0d9477687fbb6d3303d7c49 : SUCCESS in 4m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddb5904715a14a788f0d227d8005f93b : SUCCESS in 6m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c85eda5244b64bc18e3f0952f3c0ad4b : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/181890746ee94c80921882d0f2eeacb0 : SUCCESS in 48m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/38e12f6bd1b04b4dabadd9d6a49ba91e : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a6a9121ef304aadde2e0199756d523381edca94c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 09:06:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b7a9674163e9dcc4c3811fa0b55e148ac9e1bb8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:20:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f069a1b109f46a486417c2f982f716d : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9221ce77fc66415b8a00a93e8ce7d157 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff80c2e0f3340bc99f8d918b1e81a03 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56c8202c833b46cf94ec569fc4557433 : SUCCESS in 6m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7217cf57a19f4d258f3a8c5d8418619d : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/232707fbb4314b15a71765310095e8e4 : SUCCESS in 6m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10e6f2ee3e6a40a794df8f2d3b5ab4b6 : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/604582059a0f4dc7bc4c9e9c967e433b : SUCCESS in 11m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b18b345dc6e34b568f1833da5f8736dc : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3ff3fd37165341f0850f7558f6435619 : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/79d357b81e61466a953df75ea223e757 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6be7fc8a4c014a47b74855bbcb5d1702 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8809906c4d3d453f9b02c2218a87da16 : SUCCESS in 9m 13s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/11b888be5341451e95a6f4d9def23f4b : SUCCESS in 13m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6e7a54b288b3411bbccfaa5310a97865 : SUCCESS in 30m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3636040485c749b38090cb4a0e3d99c7 : SUCCESS in 14m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2d69fefa2ef146e9b0100ca4333d24a8 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff3d5f8449d948e48414008bcc52e5d5 : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/df44a85e331e45b0a8697c32d4e207ae : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b84037dc00541aea7d84e00ad34b046 : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/21b835e34aa4432f92a7e11d786ad382 : SUCCESS in 9m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dd56eb303b9438bbfeeb7d5486b3724 : SUCCESS in 58m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b63bae091fa1474e89815de2f7ca802b : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":1},{"id":"dba85f3762a446518361be8b2e8552ea9af5a66f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:20:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7186462190750bb140716579c995a72352e4f078","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 15:31:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e9d46cd66284af8a253125c6e94a785 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"802c5b21ae8dbc30410244165442eeafe5425e00","revisions":{"802c5b21ae8dbc30410244165442eeafe5425e00":{"kind":"REWORK","_number":1,"created":"2021-01-21 09:27:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/74/771774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/74/771774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/771774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/771774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/74/771774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/74/771774/1"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:27:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 09:27:17.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.5.0","message":"update constraint for ironic-lib to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: I7c4286c9cba33de85b18e585e6787ab4c3e08c8f\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I05046492f29774031012bafabd2733def97f6dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/802c5b21ae8dbc30410244165442eeafe5425e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/802c5b21ae8dbc30410244165442eeafe5425e00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
