)]}'
{"id":"openstack%2Frequirements~771864","triplet_id":"openstack%2Frequirements~master~I4b94edb2b26e9692ee3a99c8423beb972a5a57bd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4b94edb2b26e9692ee3a99c8423beb972a5a57bd","subject":"update constraint for sushy to new release 3.6.0","status":"MERGED","created":"2021-01-21 17:37:32.000000000","updated":"2021-01-22 08:42:53.000000000","submitted":"2021-01-22 08:40:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771864-new-release","meta_rev_id":"3e14dfd00e5cf64bea98d38113143d6155de0e73","_number":771864,"virtual_id_number":771864,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-22 08:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-22 04:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-21 17:49:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-22 04:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 17:49:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-21 18:23:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-22 01:51:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-22 04:14:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2a768a945ed6e2d8730156e538498d52ee34ee46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-21 17:37:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f2143fb16ae27f0277debf50380236b4bcbd7c3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-21 17:49:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"646c1528733e933ab691a57fecc8f1a9bf0977f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 18:23:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"aa3cece86feb99a13d7e92a620f48afd59e5f8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 01:51:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d2811a0dcf4d5d9d6381be2f6f3a43 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dee70095499f4a34b6c00e20dff52bc1 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/915945ee8e734744b2735489b789a866 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd7c99ca811c456fb20f8e90eadc2d86 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d13e14be6b54c84b0bc43069d77a283 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc95f18183c74c62a5bc419ad810a501 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8393e27a6df94e91a22f6f15345e1a90 : SUCCESS in 3m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b342d91c4b464a05bdfd93d614bc6e5e : SUCCESS in 8m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a37a94c2cda4464d801b3f3d59b442e8 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d370f9bc1bc8421eb9244cb57108f6af : SUCCESS in 9m 13s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4dca3a27426d49799169a2d14295036b : SUCCESS in 6m 40s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/22a63da02ec54feda2415b3d356aa91d : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d2bc679cd3284cadb1df417ce44b0825 : SUCCESS in 7m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8d6d0ee5149942f2bdaa4141f3426cdc : SUCCESS in 8m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f99478d7e2554bbe9e5588616dcf0a36 : SUCCESS in 12m 58s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/522e69bd480149669908b319f1af3dbf : SUCCESS in 26m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7b227075e1254c078fa4c418237a4ab0 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1de50d13be2343329993ee6c4b47b851 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d6e87f0a088b4aa5a9f7a4ee4cd90311 : SUCCESS in 18m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e4dbe33b98044459b4338958eea3e4af : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/50f41445a0104223bbe859be81f2a06a : SUCCESS in 9m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/157d0555e1434af387e5e2303781e760 : SUCCESS in 11m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5972dadacd8d484ca9321e61ce8119c2 : SUCCESS in 1h 00m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/691fc2bfbd094dcb84f5f41383adb474 : SUCCESS in 1h 42m 47s","accounts_in_message":[],"_revision_number":1},{"id":"99d1a79c9ca86218bb9bcae73b364d0b19e391e7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-22 04:14:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aadbe6aaf27638b1bf0e680561e8a688bcd3164d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 04:14:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e056f6e0d10806764ecefa99b0b6a891f9201bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 08:40:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00bbb1fddc840f2b84861452ead6ca7 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419bc466438942b3b9fd1cdc35b7575d : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15baf2ee19a5412f88b0dc4dcc681ab5 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22744257d3474d868f1f17a6bfec8283 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7afb92c03634c83ac3ad804b95ab305 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/74c5a3160b6242a1b7fe9be23fb509fb : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e2f44404a8804b5d9d1e99f09e886f0f : SUCCESS in 8m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b99b8dadff1344dea5ff5f2827544db4 : SUCCESS in 20m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/491954f8838b4614a9a3b80bc2b105f4 : SUCCESS in 8m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1d16ca337ab54fe39993e67c5a64daa1 : SUCCESS in 5m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9588f49d839a4a059896e44756ac8f22 : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0c90c6f363e498e9f087bc99f5793a2 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8d0c94fb9c7b44a8a50ef2493bab4753 : SUCCESS in 6m 54s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/703734cb822a4d0882a1ed9b9d445b98 : SUCCESS in 19m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f03a73b725054c5a89362ea8aa52172f : SUCCESS in 26m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b5a25f69f6146b7ab82c38f34219fdf : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1d277c56d5d649b3b4db92d5c0cb9783 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/721e49ae5f5847bd8f3e826adc559c1c : SUCCESS in 17m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0c6e543a43a346b6a96213cb66e3e97e : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de5306646c65403cb52cd51893b664bc : SUCCESS in 6m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/665872ea36f24d81ada867af9207f7a2 : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1506ac24aef471e95b1ca7b92a074b1 : SUCCESS in 1h 01m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc204c7115234f0b91b8a96eaeb8f319 : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"864494993784086a92822f4165a8d00a7122f1d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 08:40:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e14dfd00e5cf64bea98d38113143d6155de0e73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 08:42:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2c957d191e64079ba8e7b19edbdfb43 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6082ca391c412277bc08e50971abede99861a652","revisions":{"6082ca391c412277bc08e50971abede99861a652":{"kind":"REWORK","_number":1,"created":"2021-01-21 17:37:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/771864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/771864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/771864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/771864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/771864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/771864/1"}}},"commit":{"parents":[{"commit":"378d1f4adfbf8c067b8a802075b5805156170057","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378d1f4adfbf8c067b8a802075b5805156170057"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 17:37:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-21 17:37:29.000000000","tz":0},"subject":"update constraint for sushy to new release 3.6.0","message":"update constraint for sushy to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: I19814fbb51fc040a761de1d95612bb9edc637fa8\nmeta: release:Code-Review+1: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I4b94edb2b26e9692ee3a99c8423beb972a5a57bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6082ca391c412277bc08e50971abede99861a652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6082ca391c412277bc08e50971abede99861a652"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
