)]}'
{"id":"openstack%2Frequirements~772248","triplet_id":"openstack%2Frequirements~master~I37489632fb937a3ede8ffcff40a4996e716f6a31","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I37489632fb937a3ede8ffcff40a4996e716f6a31","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-25 06:21:26.000000000","updated":"2021-01-25 10:48:20.000000000","submitted":"2021-01-25 10:46:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772248-openstack/requirements/constraints/noclob","meta_rev_id":"52b7ec998334f5307aa9d561448c705354a352c1","_number":772248,"virtual_id_number":772248,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-25 10:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-25 07:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-25 07:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 07:11:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-25 08:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b45bc5459f8460a51356088772c35ab94708cc44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-25 06:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fda27c204bfa6744170addfa20773a0f304264cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-25 07:11:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f174d883ac14ea2f2a85206914a709f9e86970fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-25 07:11:50.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"6da277653e4ca23e65918d570bf17e5d2bf9bfe1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-25 07:30:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ffa343d141d24097e97f486f5b66a076db879d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 08:40:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a98b865baa849f2bd674c3b106c1742 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/565cc7ab074e45ae8a9dcc74eb361681 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/124eb5a8200746d6ac6b234426a26d15 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1025d44742024903824114d73d1f96c1 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60c51c04f09044fbb77470f8956e4ffa : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e74adbf47e404d428b913b108f51a9a3 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/17bc73cdd06e4ce3a30090715ec03076 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/669ae31cd06f489aae0b0058cdafa420 : SUCCESS in 7m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5662e1b32a7548f4965c5204938e281a : SUCCESS in 9m 20s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cf44250e41774907bb7b121ca9730bce : SUCCESS in 7m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6298bf4921934dd79e52c02aa277645e : SUCCESS in 4m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b025397210404c46a78523af505bfedd : SUCCESS in 8m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2bfc5411ebe40bd8396a77877fdc409 : SUCCESS in 6m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1543db0164234aa284cd154213dd009a : SUCCESS in 10m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6e20e19e19114ad8932bbf1c3578f90c : SUCCESS in 15m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a262ca10a034fbbbaa322b7867d4bb7 : SUCCESS in 25m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/554201ded56a4f2a9c5712063d3cdfa7 : SUCCESS in 13m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3d70b24d9a14f97a3bcf608ed702163 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/73709e9d601143b99aa5189c73c21607 : SUCCESS in 18m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d67a2778d0e24dcaa0895b1f2b3addfa : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38bc6e1dc2ef459eba36dac9d136a5c6 : SUCCESS in 7m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1557c9dc7ce7455d8e62eaa2518b5d8c : SUCCESS in 11m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cc6956574364f63b75caa06e4ff7b92 : SUCCESS in 1h 09m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3de0288581b449698205a9540eefa5c4 : SUCCESS in 1h 27m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c0b680c26e60984663761f2c2710f8f382b8d785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 08:41:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f0c21efd1e9a81be6312bb2ec9bfa8678196d97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:40:40.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"5c26f0dac376fe964f303315ebfb80a9700a2f4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:46:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e643dec76747008af3da35ebf714f9 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/467fcbe8c2e8485cab8cd1adfe7e2263 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c37fdf1720ab46709203b384fb6c5de7 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7b0703fc53e4422ad45dacfd197bf01 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/149841019c00442dab1cd9ea56c9578d : SUCCESS in 6m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4fba47d86f674af98a8dd27d15c7dd7d : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/490c69d7c3c94edc8fb0e7575992ee62 : SUCCESS in 11m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d03ed164cb9d437499b254901fcb1de6 : SUCCESS in 9m 28s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6f6307b0713c453b838126c318e45d46 : SUCCESS in 9m 37s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/13571b48d8594dd587a4ae0b4e306ee4 : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/53d7baff586e4d1f84be7516f7952201 : SUCCESS in 9m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d4b8c01463714838b1a19824a23ff17b : SUCCESS in 6m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/689b3f1bcf33421aa257600977b508f9 : SUCCESS in 6m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ce6dc37484644be19d8dd5eafe9ebb6f : SUCCESS in 14m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1284604ce6354736a66bfb17d7ed5985 : SUCCESS in 32m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3620ad24be4a40b5b891474dcbecba3a : SUCCESS in 13m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ede0fe6374848e4ae35f7ad5fe34d1e : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3373edfbfcfe44d3bee9c7f4caf1c031 : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/923196f1d16043d6803187a91d360ead : SUCCESS in 9m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6dcb823a12ca4a28a20b1827a2f3afa6 : SUCCESS in 7m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f686c08bf744da7832e03ee4d24012c : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ad310539ab44052a2b121d5b7d82e12 : SUCCESS in 47m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a40bac8f10e4cf080c6d15d46e2b99e : SUCCESS in 1h 30m 05s","accounts_in_message":[],"_revision_number":2},{"id":"4de6dc9ba70f841abe88b25663dc62b634937f84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:46:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"52b7ec998334f5307aa9d561448c705354a352c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 10:48:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3523e6cda00f43acac26e48af916d4d9 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1be19b0999c56b9c721104241c791c4054d2facf","revisions":{"33dbc75d33fb694b76b7b47f7e8d457b6607826c":{"kind":"REWORK","_number":1,"created":"2021-01-25 06:21:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/772248/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/772248/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/772248/1"}}},"commit":{"parents":[{"commit":"42a7a33cb14dc5e24bec27a3df315b5990f99084","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42a7a33cb14dc5e24bec27a3df315b5990f99084"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 06:21:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 06:21:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I37489632fb937a3ede8ffcff40a4996e716f6a31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33dbc75d33fb694b76b7b47f7e8d457b6607826c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33dbc75d33fb694b76b7b47f7e8d457b6607826c"}]},"branch":"refs/heads/master"},"1be19b0999c56b9c721104241c791c4054d2facf":{"kind":"REWORK","_number":2,"created":"2021-01-25 07:11:45.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/48/772248/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/772248/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/772248/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/772248/2"}}},"commit":{"parents":[{"commit":"42a7a33cb14dc5e24bec27a3df315b5990f99084","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42a7a33cb14dc5e24bec27a3df315b5990f99084"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 06:21:24.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-25 07:11:25.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I37489632fb937a3ede8ffcff40a4996e716f6a31\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be19b0999c56b9c721104241c791c4054d2facf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1be19b0999c56b9c721104241c791c4054d2facf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
