)]}'
{"id":"openstack%2Frequirements~772433","triplet_id":"openstack%2Frequirements~master~I45568f758c2c09dbbc5404be5ae7478d2e1ef9f1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I45568f758c2c09dbbc5404be5ae7478d2e1ef9f1","subject":"update constraint for python-mistralclient to new release 4.2.0","status":"MERGED","created":"2021-01-25 20:47:58.000000000","updated":"2021-01-26 03:55:07.000000000","submitted":"2021-01-26 03:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772433-new-release","meta_rev_id":"6203b8270dceb1f67981d2d9e2de3ec8acf43022","_number":772433,"virtual_id_number":772433,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 03:52:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-26 00:23:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-25 22:50:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-26 00:23:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 21:43:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 22:50:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-25 23:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 00:23:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d2d061b8b0e4cdf51308cde156e7dcce92588bb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-25 20:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb817fc5d35de20b9ecc13cc6eafba17e9352536","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:43:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"0ee6fcfd64846fd8eb8b013e4a4152ef00e2d52c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 22:50:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19123cadc0efd5a2443caefb38e6bfb753429888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:04:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7149236484746d5bfd224018c87ba2f : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576c22a6c97844fb927ff40829c26503 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4019100582e4ec796618d2ef636b80e : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbf693e8303840d18653bcb583b10393 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4525cba0a10a4a1bbb8e83264bedbcab : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/efc1b361717a4e00aae5b625d91848f2 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abfc91e5ee0a43a383876badfa12235b : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf4cd3464fed46e58a4c83cd7ac69e0b : SUCCESS in 8m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f35676d430cb4377bc08ba384f8deeaf : SUCCESS in 8m 16s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/79a39de48e7849f8a3d3da1a3a0b7446 : SUCCESS in 9m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ca5e7ad975894f9fa3a92faddfd77b03 : SUCCESS in 6m 55s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/073b95bdf11a4c058b4e60d443afb5a8 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/43ebc9c76cb949df991ded2aa4676b5f : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63cf9c04f8924aee9605cac6db22e4d7 : SUCCESS in 8m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/35a9b25f228e40dc922c3bc5e2d30ecc : SUCCESS in 12m 53s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/353e377c92524c64b1b3b14460f38b00 : SUCCESS in 27m 15s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6a01e8a442df42c0add800ec42523dff : SUCCESS in 13m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9704bb34ca6345548ad37b6815af5997 : SUCCESS in 5m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b4e5ec9f60546258a2b8dded4e8976d : SUCCESS in 18m 59s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c0527d15eb24279be059e373dd5bc65 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6b2c1621e44a4ce383dbcc1edcda1f09 : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/be562a94890c4f57b315a71cbd3ac080 : SUCCESS in 11m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cff2eb0b9a344a3fa76f7128c708fb09 : SUCCESS in 54m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd9222b01ca34b11b660eb7ecd1948c3 : SUCCESS in 1h 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f59807f4181ddaa3389d017c8d9bfdabbdc0d769","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 00:23:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"253b755699d82694ea9116c1fcc23272ca6b91b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:23:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"718fe4ca000aab4f203fac1d0f0303cb3020fb2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:52:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e1c1e5ca1c94e13ad5078a65bece729 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b3cba59bfe942bea446427e4db84d32 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994a9dce10504c79ad72455e8aa8d00a : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e31d403884814985b38fdffa5ce32180 : SUCCESS in 4m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2bbdd75ed77d499391f8a16f2dd77abf : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73989d8779284135987b12e924845cb7 : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/04a51c4f996845549ff2988ccc215f7d : SUCCESS in 8m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7beb78a4703640d1acf02db607a9ce94 : SUCCESS in 8m 57s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/05078ba172b84669b3a223fd90ef20ce : SUCCESS in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4e8e244edc2d4902b5692843079f7628 : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96c5477d10ff451b98510adec64ceae1 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c68dbb482f64fc8a088002d53bf9dad : SUCCESS in 6m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9d751231686c4139bb483caa93ca5530 : SUCCESS in 7m 02s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/734eeaf9f985479dafbdfb077335edb8 : SUCCESS in 12m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6d1bd20aaeff4a24a2bd20407abeab72 : SUCCESS in 25m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/648ae4caa7c847668a5c246ebbf59698 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a11155b38c82404f8392b4923876f7c3 : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/042210fd5c7144e6b80e756cff896097 : SUCCESS in 17m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b8fbe3dd23b4fdf90c1141056b662de : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2a225642f2b42eb9d43c4cb6187a21c : SUCCESS in 7m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2bd998ecd87e47f2abcc79d1a216c87e : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25cd925f709c4f90a7583fc0c355a0af : SUCCESS in 56m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe8ec725de874bcc847291d1b74815ef : SUCCESS in 1h 23m 38s","accounts_in_message":[],"_revision_number":1},{"id":"6f87b9183a5ef0b61dc166255565182057d4d38e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:52:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6203b8270dceb1f67981d2d9e2de3ec8acf43022","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:55:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7689e381b47240299872e2155790a470 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc430da8dfe60d99832a77eecfbb8cdfddf189d0","revisions":{"fc430da8dfe60d99832a77eecfbb8cdfddf189d0":{"kind":"REWORK","_number":1,"created":"2021-01-25 20:47:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/772433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/772433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/772433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/772433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/772433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/772433/1"}}},"commit":{"parents":[{"commit":"c9856f4bf4bedf60670ac659387c9995c2f8dfb8","subject":"update constraint for osc-placement to new release 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9856f4bf4bedf60670ac659387c9995c2f8dfb8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:47:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:47:55.000000000","tz":0},"subject":"update constraint for python-mistralclient to new release 4.2.0","message":"update constraint for python-mistralclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I75de9bd1b6d6a9bd2b9cca9ea349e8a489045cfb\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I45568f758c2c09dbbc5404be5ae7478d2e1ef9f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc430da8dfe60d99832a77eecfbb8cdfddf189d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fc430da8dfe60d99832a77eecfbb8cdfddf189d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
