)]}'
{"id":"openstack%2Frequirements~772434","triplet_id":"openstack%2Frequirements~master~Ib16fbff23fac5a18fefce0842dca1def1d2e8f22","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib16fbff23fac5a18fefce0842dca1def1d2e8f22","subject":"update constraint for python-muranoclient to new release 2.2.0","status":"MERGED","created":"2021-01-25 20:48:34.000000000","updated":"2021-01-26 03:53:56.000000000","submitted":"2021-01-26 03:52:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772434-new-release","meta_rev_id":"d57e8afa31c4696ae8ee7143076af9b8e1c2af0a","_number":772434,"virtual_id_number":772434,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 03:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-26 00:23:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-25 22:50:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-26 00:23:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 21:43:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 22:50:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-25 23:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 00:23:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"577a977a94ca5d4e8ff1281b5554c616ed19170c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-25 20:48:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f2f75650ef5e3b51e1fb63a359df825a47d3cc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:43:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"ba0fd5889079c4a0b626d542eafe0bcb2d01d951","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 22:50:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef34330db540f0b0779ecd3eef76939b4d6d64aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:45:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64969b650b8d43da9ec02bf64c691660 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5a676cb3d0942f093dfb9d1517a0deb : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85bca7c2bcf046beaf9a90f0e74dc8c6 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c845fdb9ce48a8a624f0286fba88b1 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d7e74554d004784977b12bb1233161c : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79121b4383214c76a331dc624ac37b56 : SUCCESS in 5m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/61e000ba8baf44409182281bce2c3ef6 : SUCCESS in 3m 52s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/222c3383a1ec4dd5ac4e17f62309cd92 : SUCCESS in 13m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4eebd347b63a4da0acd73b66bafd6fef : SUCCESS in 9m 33s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6824e5efddb1411185969cb94524ae9e : SUCCESS in 8m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/db08fbb27bb142918798a21b8d2bb5bc : SUCCESS in 5m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/e4a990335eef41ba9e41051fee0e1bea : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc94ab903e6d42d292b3d60235fffe8a : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/736932000d464eadb779a4aa51b8a5ac : SUCCESS in 8m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9b3ff1806f8d4409b6d9b06564e833f9 : SUCCESS in 17m 38s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0853bcb4b27040d384b035426cf7aaad : SUCCESS in 29m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8770beaf489748f6bef2aab3148bc43a : SUCCESS in 13m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee09d416b31c40bfa9482f92b4fa1538 : SUCCESS in 5m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd359a5d2d0140d793a58e3b994006a2 : SUCCESS in 22m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8f964dda6d8b458880d0deeab1ea46db : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39e34d5244e34240b2f7c6e1ccc23f3c : SUCCESS in 7m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/88fbdcc597d44c778db54032ffe0d6d7 : SUCCESS in 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c63d48d9d6ca481e967deda9d17f994c : SUCCESS in 1h 17m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d275766157c4ef8aa8d5eb64a9753d7 : SUCCESS in 1h 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8893035922aa49835ed2f0b95bece3fa309ded56","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 00:23:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5a2b815959b257883956818058bcf74a74e0540e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:23:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2844f87ba63e947c1355643a8267c28512c61802","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:52:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a182ed7272b34afcbc3e31350107bddc : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f420d1d5a0486887d3301d7cb89a92 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4e587982654d34bb6a0edaa805179a : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe032b3882641dea855bbb8fd605952 : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86ce549079da4272b95c38a1dc1bfde9 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ebbeaad1bb01444487f8465d6cfe5822 : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93fb00a688824aac97996de37bfebdef : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f1c8d27b25d4914aa646870702ee505 : SUCCESS in 10m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d207273f067841ef8ab815d3569a94f4 : SUCCESS in 9m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ce862a5191b9462b9590a9ce3ab464c7 : SUCCESS in 4m 59s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/081d66def9754081826d73548b891600 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b5c4cc859b6496780498a43905e82c5 : SUCCESS in 7m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3fb252a8eb3842caab34237f15ba4cc9 : SUCCESS in 6m 55s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a78b3b9dfa064859a8e005e916a9d7c3 : SUCCESS in 15m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/51bd8b182f6a490c88fcbb39d3c810fb : SUCCESS in 25m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a79484922ff445728e310b5d79ac0572 : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/535222e90fbe4b1e8c255399bd81a72e : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fccf2b74a9f4e06aeaf8133d5500a3b : SUCCESS in 18m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/dc66fe612f58411eb8f8c224efcade83 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c0e27d3cb3c487cb4b41aa55ef5d4de : SUCCESS in 6m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2735d0a14d3149729c86f56b5431abce : SUCCESS in 10m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/894e4912760a48c8a4f44fafbc3824c9 : SUCCESS in 53m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08d0938d3a36425abce676643c6bfba4 : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"604b174f684aa8048e1f82dbf815ac78518a4690","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:52:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d57e8afa31c4696ae8ee7143076af9b8e1c2af0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:53:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b4d4bb87e34551bd013f0acd627c08 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ede300ab9d27d568888bf513dd060e1c89b9120","revisions":{"0ede300ab9d27d568888bf513dd060e1c89b9120":{"kind":"REWORK","_number":1,"created":"2021-01-25 20:48:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/34/772434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/772434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/772434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/772434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/772434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/772434/1"}}},"commit":{"parents":[{"commit":"c9856f4bf4bedf60670ac659387c9995c2f8dfb8","subject":"update constraint for osc-placement to new release 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9856f4bf4bedf60670ac659387c9995c2f8dfb8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:48:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:48:29.000000000","tz":0},"subject":"update constraint for python-muranoclient to new release 2.2.0","message":"update constraint for python-muranoclient to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I905cb083064fa0663fe6436651bb58e18f982dae\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: Ib16fbff23fac5a18fefce0842dca1def1d2e8f22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ede300ab9d27d568888bf513dd060e1c89b9120"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0ede300ab9d27d568888bf513dd060e1c89b9120"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
