)]}'
{"id":"openstack%2Frequirements~772436","triplet_id":"openstack%2Frequirements~master~I5092b453bab8b3f9e0b353d439a83b60c850a721","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5092b453bab8b3f9e0b353d439a83b60c850a721","subject":"update constraint for python-designateclient to new release 4.2.0","status":"MERGED","created":"2021-01-25 20:50:21.000000000","updated":"2021-01-26 04:11:53.000000000","submitted":"2021-01-26 04:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772436-new-release","meta_rev_id":"c0eece6e304769287e45ada7f6f77fd5c6158c05","_number":772436,"virtual_id_number":772436,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 04:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-26 00:23:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-25 22:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-26 00:23:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 21:43:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-25 22:51:16.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-25 23:21:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 00:23:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"21f5bce0c2afcf13698e7b2c191c6f5626a876e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-25 20:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98c1794ec83563e0ae1b6ebfc2fdb78dcbc6190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 21:43:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"278704c372cb63032bc3aff895f1a33929a4826f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-25 22:51:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ef70bc859601600bd84e9b8dd4bba90f89a9e3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 23:21:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afd77632f3094185aff58003e898991b : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903835d5d62344f1aca9b6dffcaa5a7c : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f69aea4aa92247b0a10b8f48e36ede8d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f551cee8b7d4884b07afede54e2342c : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/077dd81b8910468f9a1e2108c5e80d81 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30ee2b2705344988a1150918734091a2 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/084210d0bdc74777a1021c638ee98b42 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2a392a3f93344f5db29e6cfd3ef75fed : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7a0e08ded4bb42b1979b988c3e21fce3 : SUCCESS in 9m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c0cad860754e4520900c1340732a0160 : SUCCESS in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/050fe39585c447e483fddf399d47f3ea : SUCCESS in 7m 16s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9716f97b4b0b4b4cb013208894774be5 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7b44bd9b84e3475c9a2e2d5dd8c2ea2f : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2cbec6c5bc714332b37bb20e59e63308 : SUCCESS in 7m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/780db27d874e464ea2f27b57eb9d6cb5 : SUCCESS in 13m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c6b401cb09d340c99f6b5cefdee5eb24 : SUCCESS in 24m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8b3a788eb8044a20bf7fb12df1b9c5a4 : SUCCESS in 10m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a9c1353d4454a1ea6b58b075bb753a2 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0ab97397e334144b3b476097e7bef25 : SUCCESS in 19m 18s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce11f300976749df8ea5cd4bd2eaa8ae : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7c861ac5cb004ed3a142c9b86065ed40 : SUCCESS in 8m 09s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/704bcddeb32643b78875ceb4a7e626e6 : SUCCESS in 10m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42587568619d4fa1827045b074bf7f24 : SUCCESS in 55m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b7d235cc36f4c8cab2b953faf96d50f : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4f6f05bd209b840f319ab860efef6b06700003b0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 00:23:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"102b4e9337e0c1c9216309af50295d8c517e74cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 00:23:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef7ed51b5d5035dfa3429513bc6fcfc23f5739e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 04:09:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1af762959794267b1791e24e673ff54 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793a6b4420d44125857a8cacb2c2246d : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d8ed1bf3a94e6e97df82cc73f7b30e : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dca02f4eb856492ea0eb4737cc6f348f : SUCCESS in 3m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d292147c1ffa4fc0a0cb1f8b5f5ce519 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/adddfc4573644bebb921cfd86965e724 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/87ca54f351ed4be697241f2855b406a2 : SUCCESS in 7m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b76e0e412c8f477ea35fda7089372111 : SUCCESS in 8m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f8c351f70cdf4acfa2509dcd886c44bc : SUCCESS in 8m 58s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d3edcf782fc2434393ebfd0a74a5bf7d : SUCCESS in 6m 35s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ae6e9001131545629fe0a9dbd44d2a24 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3aa562525afa41e8a2f07e591a9040d6 : SUCCESS in 6m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69bd5f35be874c8c9dcebf056d4c97aa : SUCCESS in 6m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/15c1f90002cc4e44883d1ebdb5e2ed9b : SUCCESS in 13m 18s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/9b4a87167738470a850f5ff91c3ce399 : SUCCESS in 25m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/be5cf7a6bbde4a418fa07dc5c5c7cac9 : SUCCESS in 11m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/993aca4336344e38bdad4e9469170aa0 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d97cb47b541a47ad894eed3ab5aede1b : SUCCESS in 16m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2d0b68060f1d4f4280934542c1e1cbd8 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/25fd3d07ccb047fab3f328dd0e5ae885 : SUCCESS in 6m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b453a3897996420299e546d372b0585d : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5168df921bf34c93a976364ae19edc02 : SUCCESS in 1h 03m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3e0cd6df690462fa3711e348d20a031 : SUCCESS in 1h 09m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ec1c7ab272f510ecfd67f98ee1fce86c971d1564","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 04:09:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c0eece6e304769287e45ada7f6f77fd5c6158c05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 04:11:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84550b9226fb4dd2acb7ce8076707b10 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bbd1f290a4ef341df1ab2aaf49896acccbd796d","revisions":{"6bbd1f290a4ef341df1ab2aaf49896acccbd796d":{"kind":"REWORK","_number":1,"created":"2021-01-25 20:50:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/772436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/36/772436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/772436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/772436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/36/772436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/36/772436/1"}}},"commit":{"parents":[{"commit":"c9856f4bf4bedf60670ac659387c9995c2f8dfb8","subject":"update constraint for osc-placement to new release 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9856f4bf4bedf60670ac659387c9995c2f8dfb8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:50:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-25 20:50:18.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 4.2.0","message":"update constraint for python-designateclient to new release 4.2.0\n\nmeta: version: 4.2.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I96b8ec21443774685cfc2102fdf29777359b8f8c\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003chberaud@redhat.com\u003e\nChange-Id: I5092b453bab8b3f9e0b353d439a83b60c850a721\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bbd1f290a4ef341df1ab2aaf49896acccbd796d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bbd1f290a4ef341df1ab2aaf49896acccbd796d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
