)]}'
{"id":"openstack%2Frequirements~772471","triplet_id":"openstack%2Frequirements~master~Ic3010d5fec2a943d526dd43f97eab0a35adeda3e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic3010d5fec2a943d526dd43f97eab0a35adeda3e","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-01-26 06:18:48.000000000","updated":"2021-01-27 02:17:19.000000000","submitted":"2021-01-27 02:15:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772471-openstack/requirements/constraints/noclob","meta_rev_id":"041c853a9ba6cfe73245e5c3f72dcb7fc697eaa4","_number":772471,"virtual_id_number":772471,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 02:15:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-26 07:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-26 07:31:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 06:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-26 07:27:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-01-26 08:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9280aca276b1a2cc76200195bbab15fc10a3db4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-26 06:18:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2a0652303f831cbc0697f0e547e562351a940e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 06:37:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57892876bca848c4ad7e8640bd556858 : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f8f82f5fc8f3d35e0c7314f9536f8557b52a4de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 07:27:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"81f045672d54a7383ec74a48504f911e059a1ac1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 07:27:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"63fd390c83456ce98fa7e57e9850b08f08717037","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 07:31:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ead28f958feebf13aa4bf177782a882ee53dd53a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 08:09:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a9b5ad474194165bb9f85af98d65b40 : SUCCESS in 37m 09s","accounts_in_message":[],"_revision_number":2},{"id":"13bc2b098627ef0105f00a4027e203b524ad7f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 08:46:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05401c9c18748168fab45d04e1f808f : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c652bad55294470596d5a78d5242a4dd : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9cb131caf7a462d8f60cf349256017d : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/576af35ba108401798ac0b0f2b317c81 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19028382be634650a94def95a2aff5f4 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e6d5b8cf53f4bc089eeff7af4707968 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/82cb694c60cc461ea9016b99f0617fd5 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2c58ed19aef64c86b8dd90b2403aa7b6 : SUCCESS in 7m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e194fbcf8cc24f4db0f5892466129e6b : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f658e1a3f62d426d92a9681a65692374 : SUCCESS in 8m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dd480223ae004bdca8ae7ca371f66d05 : SUCCESS in 4m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5407ecc9f91d41bbaaab03d0e9905ac7 : SUCCESS in 8m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e33b3ea7d3e74a1197bf339681b77604 : SUCCESS in 7m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63431ec339f34bdb879c5fce4198ddf6 : SUCCESS in 8m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8439f3c417c34099a717802b5e4a40b2 : SUCCESS in 13m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/657e0e2f6dcf49c98bccc07b3ae11da2 : SUCCESS in 25m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cd13789d6ef14cf19d5f064da16d804a : SUCCESS in 11m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/feb4ae53130d429185be6f881450f529 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/db9e1be5f06345848938dec3dd5587fb : SUCCESS in 17m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/259f757a107a40eeb7a31b7b5d9da239 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04728f68275e4d9d80665690511240c1 : SUCCESS in 6m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a77d0f4dcdfa4531b36b120d9aa03358 : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/108d4ed43aa34521b88e68a3c892d939 : SUCCESS in 1h 05m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/55848b156c8b4948bca6ba6ce58646d9 : SUCCESS in 1h 09m 15s","accounts_in_message":[],"_revision_number":2},{"id":"3da7a4ea5c0050791ee16e17531818e4e0f92dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 08:47:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca0b340f80fe56ac5b82a0d72a0e3a91f8872a28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 11:32:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/613a5493267141faac4ffd370fba7cb2 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a98c2c987a64454582630e539476fd5e : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5af42cde9f1421c8486f2a9ea02a8c8 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c931a6c84cff434689d17cb2e0478832 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b9180fdc0744251bcf5b5e4bb8fbdc7 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0cad8739b0a46fd921b7b166fa4516e : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/eb0835a36ab74294885638f4ee71ae5f : SUCCESS in 9m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a50fea988c334fea9deb3f41721fb940 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9e44b694a5b347338805f7e43b56a996 : SUCCESS in 11m 40s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ee75450545e64b2cb567bc64f5f79e6d : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a37068aa63f54fbd96dacae94be09212 : SUCCESS in 12m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d2a4444367849e6beb1eba453a891ef : SUCCESS in 7m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/81428aaee20a4fe9a5c49f7340cf3e84 : SUCCESS in 8m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83219d6f6b5144169f801f285f76b1fd : TIMED_OUT in 1h 01m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a0be6d06e072418fb4cbc5dc1d7559f3 : SUCCESS in 25m 20s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9218a5657d8f491fbaa6011bb5784ab8 : SUCCESS in 14m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e159eff5242d4ba087bf1235c8b95268 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfcd78d0a72c44dd9c176afbe3049624 : SUCCESS in 18m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/cb2929fe3b9d4358b8feeb2ad46ed144 : SUCCESS in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a246362eb9c443a0bc07b29b60e358d2 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0bdc9b68370348239a681f7ccd46b33b : SUCCESS in 10m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fb8ec6ab2b34e398022a0d9c1ee47ef : SUCCESS in 55m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6067be2d3fe94ea5b31f99f372d9c9fc : SUCCESS in 1h 24m 52s","accounts_in_message":[],"_revision_number":2},{"id":"7b05beb1c5a613275e3a33b54a063626f628195d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-26 16:32:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3d3000017ab22ec247b7d081beaa9641fd0bdeca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:10:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/256190e828f2447b9223b8ccfc70ac7c : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ed38d960cb803295e8b3824ceac6a4e35e5d971c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:01:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5e06d7b6734768b954fda35e3d8777 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e4a461ebf94989b3f7a926dd5afeb5 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1087b170fca1466db752fa3f3d9b1655 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96c36db918604ea19d4a5890b3e79813 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bfac39f26ad418a8d99d5950a52565d : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b3efe22ac4244e1c80b1cf98a80bc2fb : SUCCESS in 3m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b7b1b127aec641a98cf7658c488d8ba8 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d279164cc3844b39cc2385db9b6aff9 : SUCCESS in 17m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ecd02defe114451fbaf9ef9274f3c31a : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8a7b46db6a92412ea902f36efe1d4f5c : SUCCESS in 10m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a7394cfae4a043aab55f2f68e65cf825 : SUCCESS in 7m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ae795ba86d4e4dc9bcce2ca0a3e04e1e : SUCCESS in 10m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cad837aa299a43b48f0faac7c0950108 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c96bc63f4c984abe99ef6cf258f63a73 : SUCCESS in 7m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/7b5dd99d6ee2443ea0a3433d378bcd36 : SUCCESS in 12m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a1b02597850c415993c64e6cc88809fb : SUCCESS in 24m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/310a5390241a4c45984ffe379f1f3383 : SUCCESS in 12m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/71d544c3b4b44f51a74c8b2bc426ca7c : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f95cf3713838415c917ceebc14aabcc8 : SUCCESS in 17m 15s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/07ec196690164c68940c3f0b72e2b9d7 : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a68b50ba2d91452db401478eaf164cdc : SUCCESS in 5m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ddd8106dace4422824d35f6172c8d83 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ac62b0d96684b5d93321d37f16288bc : SUCCESS in 52m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f6153f47f6e41abb4f7c05d1431c728 : SUCCESS in 1h 49m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4581a5dea68ce0a02a451837cb62d2959085ae13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:01:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc218e6aebf08d5a2e5e9bf0a7ea3679bb6df2c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:15:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382dcbb87bb44a73bb0a93f9809b615f : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7bd96802cb4c4284f8fd6c4dc9c668 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34d2788a848f4d38a8d3fe549dafb368 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45e486aae15348c39fd7b152658566eb : SUCCESS in 3m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbc70036c03e4742a2a6fb0c7f6b2018 : SUCCESS in 3m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/803ad4b7edd94fa89012642e1c303496 : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e0b54d32b9b64142848c8d67f22310c9 : SUCCESS in 10m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/040a8afa225740b5b8a1cbc7c11df358 : SUCCESS in 8m 18s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3e606cad7b1b471189f5739afa54485d : SUCCESS in 8m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d2331a18ee6f4458a154335808be2e4f : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f36115c5fc184fdeb7beb0f203a97c5f : SUCCESS in 9m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7df1313994aa4f90b98131ec1ee3686e : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bc5f91da231a4bd49bc1b53e9ca650d9 : SUCCESS in 6m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/57bece72258343039f7c2d71040ffb34 : SUCCESS in 20m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/db7ee66d2313492d97a17a6400df6886 : SUCCESS in 25m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5d0fba3b29d44518a7bc9d053ec12c2a : SUCCESS in 13m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/087b31a15f62453aaf38333eda67bef9 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4f71e1bea2f34920899671198feff4a7 : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a7845fe7400e4cc29c129ec8919fa061 : SUCCESS in 4m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94a85ce5f43d46fcb7ded1239ec70095 : SUCCESS in 6m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea385d7683cb4072a8df793c0c2e4e8f : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ead8252a3781483a9f48562f01efc58b : SUCCESS in 54m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9322879b46144c6eb44d16fb5f050c9f : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":2},{"id":"fe31c4b8fcb36fe6118ad765250c8106e4d5dad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:15:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"041c853a9ba6cfe73245e5c3f72dcb7fc697eaa4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 02:17:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014edb19451c4c7e9f4c28d700fe9c0a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de3531732d3ab30e06add48bfb00f58c60a3a14e","revisions":{"10ab46a05c734d2876f1d52e1193c5957c414c9d":{"kind":"REWORK","_number":1,"created":"2021-01-26 06:18:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/772471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/772471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/772471/1"}}},"commit":{"parents":[{"commit":"641cffb2c420214d4b760c69a080987242618695","subject":"Merge \"update constraint for python-designateclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/641cffb2c420214d4b760c69a080987242618695"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-26 06:18:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-26 06:18:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic3010d5fec2a943d526dd43f97eab0a35adeda3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ab46a05c734d2876f1d52e1193c5957c414c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10ab46a05c734d2876f1d52e1193c5957c414c9d"}]},"branch":"refs/heads/master"},"de3531732d3ab30e06add48bfb00f58c60a3a14e":{"kind":"REWORK","_number":2,"created":"2021-01-26 07:27:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/71/772471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/772471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/772471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/772471/2"}}},"commit":{"parents":[{"commit":"641cffb2c420214d4b760c69a080987242618695","subject":"Merge \"update constraint for python-designateclient to new release 4.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/641cffb2c420214d4b760c69a080987242618695"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-26 06:18:45.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-01-26 07:27:41.000000000","tz":-360},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic3010d5fec2a943d526dd43f97eab0a35adeda3e\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3531732d3ab30e06add48bfb00f58c60a3a14e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de3531732d3ab30e06add48bfb00f58c60a3a14e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
