)]}'
{"id":"openstack%2Frequirements~772856","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I9eb6d339d728567cdb66447cbe91e332a967c1d2","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I9eb6d339d728567cdb66447cbe91e332a967c1d2","subject":"update constraint for neutron to new release 16.3.0","status":"MERGED","created":"2021-01-28 11:09:23.000000000","updated":"2021-01-28 18:14:24.000000000","submitted":"2021-01-28 18:14:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772856-new-release","meta_rev_id":"2b786863891d5a3f8b6b59d67602b00e5fae4248","_number":772856,"virtual_id_number":772856,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 18:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 16:34:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 12:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 16:34:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 12:36:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 14:36:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 16:34:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"73b74f7ab1ba1e4a1c0a9d0f17dee9e1f87bae33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 11:09:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891580a77a1ba222052867a18338e6a8caf06242","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:36:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f91cb96e41dc8125f05dbe8051ade2274b2a2418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 14:36:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0b763188c44c1fa676cb203dadc64a : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/693591488da04467ab8f0eb0d15dcfa5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ddf95ea64c14e9cac697c0413191074 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2620e5a8b9fd41e4aec09b3a42d3bd39 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b9e57ba26c76458d8f0967716e2cbc3e : SUCCESS in 4m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0f7419e79c342b0a6fe3133514f4505 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d2af65d385144299924e7a7c54ba8588 : SUCCESS in 8m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8e4e04b6d18c48dd8fd234df44db9938 : SUCCESS in 5m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3922438810ff4ebba1e4ec71a1d85741 : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29ab6d57c8bc408fbad72812bd488a9f : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/945dddef188f469f89649ad40483cbd6 : SUCCESS in 13m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a91a3e76f39445539c2f5469fc971717 : SUCCESS in 26m 12s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/012cdbda8867467d9d03bf84b467e86c : SUCCESS in 16m 46s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/1e945f617afb42aa8f0d63104fe505d8 : SUCCESS in 19m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7908f103fe964a7f8bc88d130df00f65 : SUCCESS in 19m 02s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d45cc9d12d3f4e0f9613572757630eff : SUCCESS in 5m 51s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/5d7d7aa1c533477b8b37e55024459400 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b76eca2e45094e01a6a7ce19eba6cf6c : SUCCESS in 7m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/666d1fda3d3540d39f2fae8378e52a14 : SUCCESS in 11m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c53daac1e036469a94d793aa6cc5175a : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0896f0d807c246c080bb5dfec0e9c14d : SUCCESS in 52m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/de74aab55c1b497e82e8f57c43fd6801 : SUCCESS in 1h 13m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/727605754398488d993808b9a21eeb56 : SUCCESS in 1h 25m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/8df949569ffa4e3697bed95d21238b58 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cbe7a23150fe9ca955bd03bb7812e58f23086b09","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 16:34:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efaf37abc2be2fec79ecd85daaf00a95f1a6d453","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaf754dea4824dd25433e5bdb26c7fe0e1d0311c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:14:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a20e9ba1a6d42a7a653a5ecb9bb20b2 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f8b44e2dddb42b995ea4ab340537eb8 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ff819c418af4d91b06b8a1b0d21ef49 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5f7a0975ca69495ea6af82e2fdabf8be : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4aa6a631ce53482380f12a4155b2fc9d : SUCCESS in 5m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f790ae369ff744e5a9cd2da9c29a54f2 : SUCCESS in 12m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8541d4d5fa6b449db6aca3297311da1d : SUCCESS in 11m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2ded51a2a11f46d3a0959c14ceb9a0f8 : SUCCESS in 7m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bfcc3fb710444f7886c2fea250f81db7 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9ffccf44088c40d8abd9410506f95727 : SUCCESS in 8m 53s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2aa6aac539084f66a22ea3cffe28c245 : SUCCESS in 12m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b7e40537e515403098c3248fef815136 : SUCCESS in 26m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dfdb299c58354807a343a283331a271a : SUCCESS in 14m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/3e801aca7ad04319aaa72be2393bb95b : SUCCESS in 15m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7198abdf9e74454aef10cf5ceb52d4f : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1bcc7bb852ce4b9d9394ab1099f1c20f : SUCCESS in 5m 54s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ff9ae3886f04405fa33618ea1da60ecc : SUCCESS in 6m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36443bd05c9b4e009d7d47de5f109a34 : SUCCESS in 8m 01s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d5e8fcf44ff0481abf6b5cc506eeda19 : SUCCESS in 11m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1914aff96e1f4eeb9c3fa965fad00aed : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b5a4d72ad504751864f030d86684ec8 : SUCCESS in 56m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23c891ca1f914b339a8524a79a77c5af : SUCCESS in 1h 24m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80fa8a7bdf0442b8825a4b147aae88cb : SUCCESS in 1h 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f1cc7072d9d34d309840e86ec949cdf5 : SUCCESS in 1h 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"2b786863891d5a3f8b6b59d67602b00e5fae4248","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:14:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bded58764a3186144ae8fd2cec2db2e973b7768e","revisions":{"bded58764a3186144ae8fd2cec2db2e973b7768e":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:09:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/772856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/772856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/772856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/772856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/772856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/772856/1"}}},"commit":{"parents":[{"commit":"ee9faac15785d70b264b363aa8e473d6d7362f60","subject":"Merge \"Add neutron to stable upper-constraints\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee9faac15785d70b264b363aa8e473d6d7362f60"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:09:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:09:21.000000000","tz":0},"subject":"update constraint for neutron to new release 16.3.0","message":"update constraint for neutron to new release 16.3.0\n\nmeta: version: 16.3.0\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Change-Id: I10f4e1e24486d59217a6a21d11f003c1ace73c37\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I9eb6d339d728567cdb66447cbe91e332a967c1d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bded58764a3186144ae8fd2cec2db2e973b7768e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bded58764a3186144ae8fd2cec2db2e973b7768e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
