)]}'
{"id":"openstack%2Frequirements~772859","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I311970bcca383613814a7cedc7713bd5279e4dae","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I311970bcca383613814a7cedc7713bd5279e4dae","subject":"update constraint for neutron to new release 15.3.1","status":"MERGED","created":"2021-01-28 11:10:16.000000000","updated":"2021-01-28 18:31:05.000000000","submitted":"2021-01-28 18:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772859-new-release","meta_rev_id":"8e7ddd94efda417ae1b5d882c5f6d19ea09246a3","_number":772859,"virtual_id_number":772859,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 18:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-01-28 16:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-01-28 12:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-01-28 16:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-28 12:45:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-01-28 16:34:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 16:39:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dd204932150e451bef87a9e4eb0d651d95a564b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-01-28 11:10:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a81ceedd86eae94db97b8667cb5170b0afe7d93e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-01-28 12:45:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27bd6b35267dc985c1412bac163e1de2cd75dc05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:34:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130ab54fd04549a8920d5599ca60a10b : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bda85e5628eb492c90531673eb16b7b1 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3fd3e224ea64a4b87c7aadcdf914e14 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac6c86661a464e9694af5702cd7b2064 : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/676ffeea92f34e5fa5f562d8be279f2a : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edbb2d08f4b24a3eade4242ccd48d43c : SUCCESS in 6m 02s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/76d200e9aa63468f940e4c5efe6505c7 : SUCCESS in 11m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3602707209024ac3b81a21f0d9bf29ac : SUCCESS in 15m 11s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4833e1bdb6494fdea83517c850f234d3 : SUCCESS in 11m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b754e072944d4708b8540177066beb50 : SUCCESS in 8m 31s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/411aa11e445d42bf8321da793a5634ab : SUCCESS in 5m 22s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5c502c488ddc4beda4e5a43f1c39dd02 : SUCCESS in 5m 01s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/4240f471dabf4a40a27e501bc4725983 : SUCCESS in 10m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/20bb39ac763648f3b9572935a313688e : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/453c34f53d054a29b8e5ce3c3b0c47c3 : SUCCESS in 6m 18s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/a5bfbe2cb6af42fbb65bc6f8894f4996 : SUCCESS in 12m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/099087bcf0014f97b2fe2652f7a40c0a : SUCCESS in 12m 16s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/50746d31ff9a44e9930653b8cbf13f83 : SUCCESS in 21m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/99ebe2a45b3e4759b95d31cd6ca6bea8 : SUCCESS in 21m 30s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/f8b900d2436c49eebc044be527244cb2 : SUCCESS in 12m 25s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6c87747144ca4097b6e695c7c5fed367 : SUCCESS in 13m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a8052b7f882347cea309678de21e2924 : SUCCESS in 15m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ae55979e06744e4af81b380c3f26a37 : SUCCESS in 19m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7605ac1b3b8748d7b7d7d6657f6fbaf9 : SUCCESS in 5m 41s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/207f74bc86124ba0b2aca396fc96551d : SUCCESS in 8m 45s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c5f709f41a76429184bb2e8242504540 : SUCCESS in 11m 19s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5b776fc6fc142a390b1e48b782d2494 : SUCCESS in 1h 08m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/fd70895523774945b48d8d23281dbdd4 : SUCCESS in 1h 38m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/64c4d6c3ebb0402fbc1461dc56299774 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f203005c7e6d301bb093048ccee527a645e50da6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-01-28 16:39:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d4f71d9700434fa2b37e6528cba080336095de4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 16:40:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"076283b0535bae5fbea54e44b16b6d3411e0846f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:31:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de7d63a4ee347e4ae0a118c2333108e : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8c31121fbd84ec4b34b2b696ec68d34 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5417db2cf51425d83c8033209c973cd : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adbe8626227f4e22bd5ca2b0c0da4890 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a1ddd95a963448b88c54cc818a975a3 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/801a2964d81b453a8988a5adef466a6c : SUCCESS in 4m 17s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/cfc6b994a8f2467cbc900bda35f6dec2 : SUCCESS in 11m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1b79a19aaf4e4e389c2bd08624e8b872 : SUCCESS in 14m 53s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/28ef2496e4c54d15a1e229bce67f4393 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d58efcca7cc84a16a0d069e0baea7a42 : SUCCESS in 10m 16s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/c2345d6b1f71488582fbd8a8c76f806e : SUCCESS in 6m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/381881df4b3141f18194a026f96b596a : SUCCESS in 5m 45s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/ea0dabc5e2a346f997af8bde9e7a84f7 : SUCCESS in 12m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/abd0948d4b03467d963120ff2c31fa12 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/593afd777c1c49f58e5edae3a955a12a : SUCCESS in 9m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c9bf103aa51b40e68ff702a6910c4509 : SUCCESS in 14m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e54e46f9054641a098ab476daf0b0a36 : SUCCESS in 12m 46s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/579c4c5c48854f89a716ec38e48db12b : SUCCESS in 26m 56s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/eb0cb965f35b423c82c2bbabce4d8d98 : SUCCESS in 22m 33s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/2ea618faf2eb4acc85f23a232e0b8cc8 : SUCCESS in 16m 21s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/24d7baf029db4f78ac82e14a0f25831a : SUCCESS in 16m 38s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/26fa16b2c5a24e7785f9b60c5e102d30 : SUCCESS in 14m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ac5499f9845a467eac1055e7cc7b5370 : SUCCESS in 24m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0866645ac4634bd29b7257703b00db70 : SUCCESS in 5m 56s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2bf0413c99ea485f8e4728b702741950 : SUCCESS in 9m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/455c12f9648642c0bcdc79395eae7de3 : SUCCESS in 9m 56s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f17daaaf14f2445485b37fb5df43a86d : SUCCESS in 1h 15m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0b712a1e2d664dc6bc5bb9d527f672d8 : SUCCESS in 1h 39m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f1e419e962f14b6e990c149780cac78a : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8e7ddd94efda417ae1b5d882c5f6d19ea09246a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 18:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38701843b9aef5718239bbb62697e336fe1465eb","revisions":{"38701843b9aef5718239bbb62697e336fe1465eb":{"kind":"REWORK","_number":1,"created":"2021-01-28 11:10:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/772859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/772859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/772859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/772859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/772859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/772859/1"}}},"commit":{"parents":[{"commit":"453513c0f8619357a8fb1b624d898d4cff819eb0","subject":"update constraint for castellan to new release 1.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/453513c0f8619357a8fb1b624d898d4cff819eb0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:10:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-01-28 11:10:11.000000000","tz":0},"subject":"update constraint for neutron to new release 15.3.1","message":"update constraint for neutron to new release 15.3.1\n\nmeta: version: 15.3.1\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Change-Id: I10f4e1e24486d59217a6a21d11f003c1ace73c37\nmeta: release:Code-Review+2: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I311970bcca383613814a7cedc7713bd5279e4dae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38701843b9aef5718239bbb62697e336fe1465eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38701843b9aef5718239bbb62697e336fe1465eb"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
